)]}'
{"id":"openstack%2Fi18n~893589","triplet_id":"openstack%2Fi18n~master~Id0ca964786a4a9cf2287b008d845f4c3a1a22e77","project":"openstack/i18n","branch":"master","topic":"doc-i18n-release","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2023-09-19 07:58:16.000000000","reason":"Change was submitted"},"841":{"account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"last_update":"2023-09-12 01:57:21.000000000","reason":"\u003cGERRIT_ACCOUNT_841\u003e replied on the change","reason_account":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}}},"hashtags":[],"change_id":"Id0ca964786a4a9cf2287b008d845f4c3a1a22e77","subject":"[Docs] Update release management part","status":"MERGED","created":"2023-09-02 16:50:09.000000000","updated":"2023-09-25 08:44:52.000000000","submitted":"2023-09-19 07:58:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"893589-doc-i18n-release","meta_rev_id":"efbaa5ff5f67ae83dd86597c6d36d1a86ef7e8aa","_number":893589,"virtual_id_number":893589,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-19 07:58:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"recommended":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2023-09-25 08:44:52.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2023-09-12 01:57:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2023-09-19 07:47:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-02 16:57:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-07 02:59:23.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2023-09-22 16:29:38.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"04d6ebc0570ab0947197b72fff7fc36bc04ff729","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-09-02 16:50:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"954f1d0c05b39870d78e2b827980ea46c77494b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-02 16:57:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/73dc2d31198b4530ae890a84b4858361\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/debc75470fab4586a0167ff3b9daef98 : FAILURE in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0baaeabf4494ac2b0f344898d092add : FAILURE in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"b2bbf2a429cd1333b3668d63a0c8a0222fe4272a","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-09-07 02:59:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e44bd12ef969373040a7092b3a606193a7c9e47c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-09-12 01:31:09.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"88c7d62586c3d3c385aa7f375be54c500b36f723","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-09-12 01:31:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d6cd2b22979fb2c95a60fcf0a0f8947a0c6e4021","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 01:42:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e2837c7132a4510b81b902c27aa9464\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84856a6578f54cb0a74419597188e901 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dfd347407dd4dee9c910d735fcc44c9 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":2},{"id":"2051b3a57d4d4619414fb715a7df46963023007f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-09-12 01:57:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"22b1e3fedf3685279fa456b3918d36fc2bc944e9","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-09-19 07:47:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"02b6f7e8cca67a8aca1e57c06b4b6b92e63f185f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-09-19 07:47:13.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0448638f5aa62e52bfb247dfe1b44d66509a620e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 07:47:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7a615caec7c267da28cd9d989a1ad5f528538dc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 07:58:16.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0de116cfb6f4abfab89d1a90c5cf234\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67858c7454294788b57939f939beefb4 : SUCCESS in 9m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48b46deadf9744fcbd05c31972ee17b1 : SUCCESS in 3m 37s","accounts_in_message":[],"_revision_number":2},{"id":"208d057711d3d40b9448b235b55e10ca8b89e193","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 07:58:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"252783f121e2a06fc6cbaf35c7a02a9b8d550a0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 07:59:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee528d736c34c6d8422713a28bd606e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad86c5609da24e33ad1018df080a7abc : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2},{"id":"efbaa5ff5f67ae83dd86597c6d36d1a86ef7e8aa","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2023-09-25 08:44:52.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"990d2bbe9dec8ffa2698f7d057905b6043e85c12","revisions":{"f74d66784f5189853351481963518257209599ef":{"kind":"REWORK","_number":1,"created":"2023-09-02 16:50:09.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/89/893589/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/89/893589/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/89/893589/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/89/893589/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/89/893589/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/89/893589/1"}}},"commit":{"parents":[{"commit":"4e8bc2dd50a595da5855463015d79d125b800edc","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/4e8bc2dd50a595da5855463015d79d125b800edc"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-09-02 16:45:52.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-09-02 16:45:52.000000000","tz":540},"subject":"[Docs] Update release management part","message":"[Docs] Update release management part\n\n- Reflects that now stable versions are created after release\n  without early creation of the versions and merging to master\n- cycle-with-milestones has been replaced by cycle-with-rc model\n- Horizon is now following cycle-with-intermediary.\n  Let\u0027s follow the same description as project-team-guide/release-management.\n- Replace a term: from Forum/Design Summit to PTG\n\nChange-Id: Id0ca964786a4a9cf2287b008d845f4c3a1a22e77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/f74d66784f5189853351481963518257209599ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/f74d66784f5189853351481963518257209599ef"}]},"branch":"refs/heads/master"},"990d2bbe9dec8ffa2698f7d057905b6043e85c12":{"kind":"REWORK","_number":2,"created":"2023-09-12 01:31:09.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/89/893589/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/89/893589/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/89/893589/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/89/893589/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/89/893589/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/89/893589/2"}}},"commit":{"parents":[{"commit":"5a846dbd436b23668b0bc4a30ad730f095c88e12","subject":"Recover -xe bash option in generatepot.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/5a846dbd436b23668b0bc4a30ad730f095c88e12"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-09-02 16:45:52.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-09-12 01:30:04.000000000","tz":540},"subject":"[Docs] Update release management part","message":"[Docs] Update release management part\n\n- Reflects that now stable versions are created after release\n  without early creation of the versions and merging to master\n- cycle-with-milestones has been replaced by cycle-with-rc model\n- Horizon is now following cycle-with-intermediary.\n  Let\u0027s follow the same description as project-team-guide/release-management.\n- Replace a term: from Forum/Design Summit to PTG\n\nChange-Id: Id0ca964786a4a9cf2287b008d845f4c3a1a22e77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/990d2bbe9dec8ffa2698f7d057905b6043e85c12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/990d2bbe9dec8ffa2698f7d057905b6043e85c12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
