)]}'
{"id":"openstack%2Fi18n~895545","triplet_id":"openstack%2Fi18n~master~I090cceba5589ec8232b9d86a8f01f364c8d15e91","project":"openstack/i18n","branch":"master","hashtags":[],"change_id":"I090cceba5589ec8232b9d86a8f01f364c8d15e91","subject":"migrate zanata_stats to weblate_stats","status":"ABANDONED","created":"2023-09-17 15:37:30.000000000","updated":"2023-10-27 18:38:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"25c8bc1dc9e96f5f12891a5fbc3b2d3e5c2e3784","_number":895545,"virtual_id_number":895545,"owner":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-17 15:58:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-09-17 15:58:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"183544b5acd0c8ba9899c96ac7b50038d9028d5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-17 15:37:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e35ab0c070b74efbe0ad1dac1c613110ce980f13","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-17 15:39:59.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ef410f3f4f42deb5ea7870f30701d2f893e648ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-17 15:58:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a0cd30e63554ac5bc521c9fc4c1b025\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08cdc3e6b472461ca5d6ce80c3e40634 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ff19c8598c041a3892cabb347deb3a6 : FAILURE in 2m 37s\n\nWarning:\n  Comments left for invalid file tools/weblate/weblate_records.py\n  Comments left for invalid file tools/weblate/weblate_stats.py","accounts_in_message":[],"_revision_number":1},{"id":"2891a0b16ae714be08c3b2e7dc4b7db99f763edf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-17 16:00:40.000000000","message":"Abandoned\n\nduplicated commit in:\n\nhttps://review.opendev.org/c/openstack/i18n/+/895544","accounts_in_message":[],"_revision_number":1},{"id":"feb5fb7b9f896284f8d6a90d304034e0cc159405","tag":"autogenerated:gerrit:restore","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-23 09:08:46.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"9503942234b8d6e3d95847e7b0f3df7a4bc055ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 09:20:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0bfc0ce4e5da469c9159d49be795d76b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1217a983f604a3ebed3a26132bde5ab : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97c25100e67f42c2a5888570bce710b3 : FAILURE in 3m 14s\n\nWarning:\n  Comments left for invalid file tools/weblate/weblate_records.py\n  Comments left for invalid file tools/weblate/weblate_stats.py","accounts_in_message":[],"_revision_number":1},{"id":"25c8bc1dc9e96f5f12891a5fbc3b2d3e5c2e3784","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-10-27 18:38:18.000000000","message":"Abandoned\n\nDuplicate with https://review.opendev.org/c/openstack/i18n/+/896306","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"246feb76ebd01f164308e525e6cac52c59085c44","revisions":{"246feb76ebd01f164308e525e6cac52c59085c44":{"kind":"REWORK","_number":1,"created":"2023-09-17 15:37:30.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/45/895545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/45/895545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/45/895545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/45/895545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/45/895545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/45/895545/1"}}},"commit":{"parents":[{"commit":"6f50121677ac098c0eca3228d2d0769c4e2a85ce","subject":"migrate zanata_stats to weblate_stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/6f50121677ac098c0eca3228d2d0769c4e2a85ce"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-17 15:36:47.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-17 15:36:47.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nstory: 2010915\ntast: 48768\nChange-Id: I090cceba5589ec8232b9d86a8f01f364c8d15e91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/246feb76ebd01f164308e525e6cac52c59085c44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/246feb76ebd01f164308e525e6cac52c59085c44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
