)]}'
{"id":"openstack%2Fi18n~895763","triplet_id":"openstack%2Fi18n~master~I8d1a2393cc50541a29bd9581c4ac2574b9dc9813","project":"openstack/i18n","branch":"master","attention_set":{},"removed_from_attention_set":{"36160":{"account":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"last_update":"2024-01-16 15:51:13.000000000","reason":"Change was submitted"},"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2024-01-16 15:51:13.000000000","reason":"Change was submitted"},"23279":{"account":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"last_update":"2023-11-04 05:27:04.000000000","reason":"\u003cGERRIT_ACCOUNT_36160\u003e replied on the change","reason_account":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"}}},"hashtags":[],"change_id":"I8d1a2393cc50541a29bd9581c4ac2574b9dc9813","subject":"Add weblate users","status":"MERGED","created":"2023-09-19 04:35:30.000000000","updated":"2024-01-16 15:52:35.000000000","submitted":"2024-01-16 15:51:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"895763","meta_rev_id":"52daa82a6c87a753e8123ba922b7d27500fdf89b","_number":895763,"virtual_id_number":895763,"owner":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-16 15:51:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-16 15:28:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-16 15:28:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"CC":[{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},{"_account_id":36196,"name":"Nayeon Kim","display_name":"nayeonkim","email":"nyn2265@gmail.com","username":"nykim"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-19 04:45:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-22 16:20:30.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2023-11-07 14:29:45.000000000","updated_by":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"reviewer":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"state":"CC"},{"updated":"2023-11-07 14:29:45.000000000","updated_by":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"reviewer":{"_account_id":36196,"name":"Nayeon Kim","display_name":"nayeonkim","email":"nyn2265@gmail.com","username":"nykim"},"state":"CC"},{"updated":"2024-01-16 15:28:41.000000000","updated_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"reviewer":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"state":"REVIEWER"}],"messages":[{"id":"87797ec6f1dbde174be8a98d1ae44c4328161312","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"date":"2023-09-19 04:35:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd854ee72a890996704659e0d3019395177163dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-19 04:45:07.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3ac8d34853d4633b2b059ef343db423\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84576b78b0464db5b99f370983111fe9 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b26d2c00c09d43b99c77f800656aed52 : FAILURE in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"52b4863f5c0f6bf6a3d65eb922f4cba227da9031","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"date":"2023-09-20 11:57:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"556493d25917c52aba569dc9eb3342915d92e066","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-20 12:10:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b1fd60f739548f8b7eb25223fb21dbf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d1172957515498f921aeed6684e4218 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b1f77f05fab4b74927f5219f3ab93f3 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":2},{"id":"86a4928f669dcf9802131ee3f15dbc0d7c90121f","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-09-22 16:20:30.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2583b0ebaf43f9caeafa11001260e911fe956bb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2023-10-25 14:01:04.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"96623840e8d0b814d989df45a8db8aa006b33ece","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 14:09:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66a2d05b14d8463da5680fca6c3f1b65\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7512cd17ecb14faeb36ec1f220553cb2 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c1e279bd7fb46b793f3ff33ba88113a : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":3},{"id":"da95fa6defbeca54e633a408ae7392c3c2142f9b","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-10-27 18:50:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"72126cd79d9be69f914c070adb6923c2013b165e","author":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"date":"2023-11-04 05:19:55.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e1c5857b208d244e0afb0f2aef45491f4ee653bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"date":"2023-11-04 05:24:30.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"800493dfedfa68ce24e103ebd71b8183d78ade05","author":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"date":"2023-11-04 05:27:04.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"aa4c00dc58c0dbec295361385c837b42c5f5308f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-04 05:32:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57d3283141e84661b55464ffca1294b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d1d87e1b0c43c4b7acf0d99a153ec1 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849a559f7d494ea1aeb960a93e14735f : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":4},{"id":"86c79b862d600cb1dca1b7b205f80f9f607a99d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"date":"2023-11-04 05:33:47.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"eddf266814b5c1d61a844d08cf95745c86b7cd86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-04 05:45:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c63609d9ca84ae3bbc4222b6d0e2a2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96e92daad982418fb73e59f52b5be9fd : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4c87e89d1494619b9de601a6b497a99 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":5},{"id":"c541c099da550cccbd0f21873842b8c06c6c1026","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"date":"2023-11-07 14:21:59.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"380667b2a214fbe5de9645c5b5daaa2d446c9405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 14:29:18.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4a9b76fbec749329181cebccdd17640\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1ca088b11024a94a76a7712aaa15f18 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f820c5c8945140468f37e936c466bf69 : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":6},{"id":"100e763b9d3b462d3f887377c3e72fa7d8719042","author":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"date":"2023-11-07 14:29:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f6880afb99458dc3b4350298d90c50f36301d732","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-11-10 15:49:16.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"27cfd60708234bf6657703d625c81215d0f52904","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"date":"2023-11-13 06:50:08.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"70a3846b66d9962fda975e53d67abc93513c70b3","author":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"date":"2023-11-13 06:52:29.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"2e0808a03e688cbcddb62a15f7223bfc7b35a10e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 06:57:15.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dbf3023e4fd4b0d875ed1273b973694\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e3e61314b74256bea59d15246eb7ab : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59c133f59a3f4fada9bc4dfbe1131c94 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":7},{"id":"99cdd37dfa457e59ca9916b41cfd1cd2ad3a504f","author":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"date":"2023-11-14 09:42:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f4732e349d30c2c04745d0e4c2eff1468381dcc1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"date":"2023-11-28 06:46:26.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"0979de73ee65b0cf7f21cc4f58c854fb1efb7973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 06:55:22.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c063523bd3d40729581a57f0832d4ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/457859f7ad72447f98740a7f629b89de : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0039a6590d0d4cd3921e0809e6e69788 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":8},{"id":"ce86de46e4a19610759eb6981f81429ab1637471","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2024-01-16 15:28:41.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a02ef6e8c3e5b4312f9a725be4f7deb73079d8af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:29:03.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"bd944cb759b0510d304e5ef7bd373700578159d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:51:13.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62dfbda62f584b63a7b921ad18b3d5ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0af62db80ae4e8f95802e69ad4b7609 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7bdef5d9cd04104a8d4f20bda798f68 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":8},{"id":"7f909c4a429fb9bdcca056b2441b11336b28bc12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:51:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"52daa82a6c87a753e8123ba922b7d27500fdf89b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-16 15:52:35.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a45b0eca0cf94c158b6d3efc652b7077\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f2abdaf65c04dbbbd382e5aface09fe : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"02b0574557982bb723abc932acef826857d96e17","revisions":{"1871ffd4ddf5f80fcf23d22a2df867a18c284a62":{"kind":"REWORK","_number":1,"created":"2023-09-19 04:35:30.000000000","uploader":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"ref":"refs/changes/63/895763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/63/895763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/63/895763/1"}}},"commit":{"parents":[{"commit":"c9c82d4f3e36a1164b195d9f690c7e536cebbc9a","subject":"Sync translation_team.yaml with Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/c9c82d4f3e36a1164b195d9f690c7e536cebbc9a"}]}],"author":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-09-19 04:32:59.000000000","tz":540},"committer":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-09-19 04:32:59.000000000","tz":540},"subject":"Add weblate users","message":"Add weblate users\n\nAdded weblate API codes to get ready for\nZanata to Weblate migration\n\nRequires WeblateUtils.py from #893361\n\nRelated-Bug: #2033946\nChange-Id: I8d1a2393cc50541a29bd9581c4ac2574b9dc9813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/1871ffd4ddf5f80fcf23d22a2df867a18c284a62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/1871ffd4ddf5f80fcf23d22a2df867a18c284a62"}]},"branch":"refs/heads/master"},"742a453dc3ad572cf1b7e67738ee3ebceaa0abe9":{"kind":"REWORK","_number":2,"created":"2023-09-20 11:57:59.000000000","uploader":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"ref":"refs/changes/63/895763/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/63/895763/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/63/895763/2"}}},"commit":{"parents":[{"commit":"c9c82d4f3e36a1164b195d9f690c7e536cebbc9a","subject":"Sync translation_team.yaml with Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/c9c82d4f3e36a1164b195d9f690c7e536cebbc9a"}]}],"author":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-09-19 04:32:59.000000000","tz":540},"committer":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-09-20 11:55:03.000000000","tz":540},"subject":"Add weblate users","message":"Add weblate users\n\nfix Zuul build failure\n\nAdded weblate API codes to get ready for\nZanata to Weblate migration\n\nRequires WeblateUtils.py from #893361\n\nRelated-Bug: #2033946\nChange-Id: I8d1a2393cc50541a29bd9581c4ac2574b9dc9813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/742a453dc3ad572cf1b7e67738ee3ebceaa0abe9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/742a453dc3ad572cf1b7e67738ee3ebceaa0abe9"}]},"branch":"refs/heads/master"},"8b4dee713acf258c4451155f7c9a255f47cdc0a9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2023-10-25 14:01:04.000000000","uploader":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"ref":"refs/changes/63/895763/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/63/895763/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/63/895763/3"}}},"commit":{"parents":[{"commit":"c9c82d4f3e36a1164b195d9f690c7e536cebbc9a","subject":"Sync translation_team.yaml with Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/c9c82d4f3e36a1164b195d9f690c7e536cebbc9a"}]}],"author":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-09-19 04:32:59.000000000","tz":540},"committer":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2023-10-25 14:01:04.000000000","tz":0},"subject":"Add weblate users","message":"Add weblate users\n\nfix Zuul build failure\n\nAdded weblate API codes to get ready for\nZanata to Weblate migration\n\nRequires WeblateUtils.py from #893361\n\nImplements: blueprint weblate-tools\nChange-Id: I8d1a2393cc50541a29bd9581c4ac2574b9dc9813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/8b4dee713acf258c4451155f7c9a255f47cdc0a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/8b4dee713acf258c4451155f7c9a255f47cdc0a9"}]},"branch":"refs/heads/master","description":"Edit commit message"},"d47defba419b3f07aef984578b8b2245598c7a01":{"kind":"REWORK","_number":4,"created":"2023-11-04 05:24:30.000000000","uploader":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"ref":"refs/changes/63/895763/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/63/895763/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/63/895763/4"}}},"commit":{"parents":[{"commit":"c9c82d4f3e36a1164b195d9f690c7e536cebbc9a","subject":"Sync translation_team.yaml with Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/c9c82d4f3e36a1164b195d9f690c7e536cebbc9a"}]}],"author":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-09-19 04:32:59.000000000","tz":540},"committer":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-11-04 05:24:30.000000000","tz":0},"subject":"Add weblate users","message":"Add weblate users\n\nfix Zuul build failure\n\nAdded weblate API codes to get ready for\nZanata to Weblate migration\n\nRequires WeblateUtils.py from #893361\n\nImplements: blueprint weblate-tools\nChange-Id: I8d1a2393cc50541a29bd9581c4ac2574b9dc9813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/d47defba419b3f07aef984578b8b2245598c7a01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/d47defba419b3f07aef984578b8b2245598c7a01"}]},"branch":"refs/heads/master"},"3a7ecd45e629c3d66fc560ff396281f171c49b00":{"kind":"REWORK","_number":5,"created":"2023-11-04 05:33:47.000000000","uploader":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"ref":"refs/changes/63/895763/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/63/895763/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/63/895763/5"}}},"commit":{"parents":[{"commit":"c9c82d4f3e36a1164b195d9f690c7e536cebbc9a","subject":"Sync translation_team.yaml with Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/c9c82d4f3e36a1164b195d9f690c7e536cebbc9a"}]}],"author":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-09-19 04:32:59.000000000","tz":540},"committer":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-11-04 05:33:47.000000000","tz":0},"subject":"Add weblate users","message":"Add weblate users\n\nfix Zuul build failure\n\nAdded weblate API codes to get ready for\nZanata to Weblate migration\n\nRequires WeblateUtils.py from #893361\n\nImplements: blueprint weblate-tools\nChange-Id: I8d1a2393cc50541a29bd9581c4ac2574b9dc9813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/3a7ecd45e629c3d66fc560ff396281f171c49b00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/3a7ecd45e629c3d66fc560ff396281f171c49b00"}]},"branch":"refs/heads/master"},"2058b7fc8d7e6866d26e45ee600954147dc56b38":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-11-07 14:21:59.000000000","uploader":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"ref":"refs/changes/63/895763/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/63/895763/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/63/895763/6"}}},"commit":{"parents":[{"commit":"d45968f182f7aee525d9117d33a108e902e48c39","subject":"Add weblate_utils.py to work with Weblate REST API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/d45968f182f7aee525d9117d33a108e902e48c39"}]}],"author":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-09-19 04:32:59.000000000","tz":540},"committer":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-11-07 14:21:59.000000000","tz":0},"subject":"Add weblate users","message":"Add weblate users\n\nfix Zuul build failure\n\nAdded weblate API codes to get ready for\nZanata to Weblate migration\n\nRequires WeblateUtils.py from #893361\n\nImplements: blueprint weblate-tools\nChange-Id: I8d1a2393cc50541a29bd9581c4ac2574b9dc9813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/2058b7fc8d7e6866d26e45ee600954147dc56b38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/2058b7fc8d7e6866d26e45ee600954147dc56b38"}]},"branch":"refs/heads/master","description":"Rebase"},"df50fba19c56159d2c5bee3bdd5356ec81a77e1b":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-11-13 06:50:08.000000000","uploader":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"ref":"refs/changes/63/895763/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/63/895763/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/63/895763/7"}}},"commit":{"parents":[{"commit":"d45968f182f7aee525d9117d33a108e902e48c39","subject":"Add weblate_utils.py to work with Weblate REST API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/d45968f182f7aee525d9117d33a108e902e48c39"}]}],"author":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-09-19 04:32:59.000000000","tz":540},"committer":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-11-13 06:50:08.000000000","tz":0},"subject":"Add weblate users","message":"Add weblate users\n\n\nAdded weblate API codes to get ready for\nZanata to Weblate migration\n\nImplements: blueprint weblate-tools\nChange-Id: I8d1a2393cc50541a29bd9581c4ac2574b9dc9813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/df50fba19c56159d2c5bee3bdd5356ec81a77e1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/df50fba19c56159d2c5bee3bdd5356ec81a77e1b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"02b0574557982bb723abc932acef826857d96e17":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-11-28 06:46:26.000000000","uploader":{"_account_id":36160,"name":"Sion Shin","display_name":"sion","email":"kmicety1@gmail.com","username":"sion"},"ref":"refs/changes/63/895763/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/63/895763/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/63/895763/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/63/895763/8"}}},"commit":{"parents":[{"commit":"d5d64997abe918b76ff03c44d402ed14d5e18d28","subject":"Merge \"generatepot.sh: Drop UUID filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/d5d64997abe918b76ff03c44d402ed14d5e18d28"}]}],"author":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-09-19 04:32:59.000000000","tz":540},"committer":{"name":"Sion Shin","email":"kmicety1@gmail.com","date":"2023-11-28 06:46:26.000000000","tz":0},"subject":"Add weblate users","message":"Add weblate users\n\n\nAdded weblate API codes to get ready for\nZanata to Weblate migration\n\nImplements: blueprint weblate-tools\nChange-Id: I8d1a2393cc50541a29bd9581c4ac2574b9dc9813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/02b0574557982bb723abc932acef826857d96e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/02b0574557982bb723abc932acef826857d96e17"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
