)]}'
{"id":"openstack%2Fi18n~896306","triplet_id":"openstack%2Fi18n~master~Ie196e43f418e730d6fee7f0a107d198e7daacbc4","project":"openstack/i18n","branch":"master","attention_set":{},"removed_from_attention_set":{"36162":{"account":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"last_update":"2024-05-22 14:15:48.000000000","reason":"Change was submitted"},"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2024-05-22 14:15:48.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-05-22 14:09:47.000000000","reason":"\u003cGERRIT_ACCOUNT_5314\u003e replied on the change","reason_account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2024-05-22 14:07:06.000000000","reason":"\u003cGERRIT_ACCOUNT_14482\u003e replied on the change","reason_account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}},"23279":{"account":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"last_update":"2024-05-22 14:15:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie196e43f418e730d6fee7f0a107d198e7daacbc4","subject":"migrate zanata_stats to weblate_stats","status":"MERGED","created":"2023-09-23 09:09:17.000000000","updated":"2024-05-22 14:17:05.000000000","submitted":"2024-05-22 14:15:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"896306","meta_rev_id":"d3d154e6787629c63aef92fa41059122365d4eb3","_number":896306,"virtual_id_number":896306,"owner":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-22 14:15:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":2,"date":"2024-05-22 14:07:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-22 14:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2024-05-22 14:07:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-23 09:20:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-24 14:41:28.000000000","updated_by":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"CC"},{"updated":"2023-09-24 14:41:28.000000000","updated_by":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"reviewer":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"state":"CC"},{"updated":"2023-10-27 18:45:21.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2024-02-24 03:38:13.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2024-03-19 13:17:32.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-05-22 14:00:28.000000000","updated_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"reviewer":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"state":"REVIEWER"}],"messages":[{"id":"409ab109ac6b2d87ac5fd98ad378ba08df06f172","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-23 09:09:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1646d93e6263a878320101085fa651db9012660","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-23 09:20:34.000000000","message":"Patch Set 1: Verified-1\n\n(137 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d949b0e331a644b5a78bf0d10d0983b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdb4293c923c45da8251ba0a8a2c0fd5 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd28acf898664b7eb8d83cc6e28d0f67 : FAILURE in 2m 30s","accounts_in_message":[],"_revision_number":1},{"id":"963f2e0cc13782f7ec9da700edf83f5cd3caa3f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-24 04:41:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d92ee70ca43cdf4fbbf27fd891591425c3db23dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-24 04:44:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"351f6e42478881a0a4371695c8a5f9e404a3da11","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-24 04:49:37.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"ae8faa72efb23c45cca270c51d55507c29fc4b66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-24 04:50:21.000000000","message":"Patch Set 3: Verified-1\n\n(115 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ac7381b45d54fbf90d3d7a947e9555b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5dc4f5a9ac74ba7a2b28541ac164ab4 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b7e12e060f941e39782d4cf4277282b : FAILURE in 2m 33s","accounts_in_message":[],"_revision_number":3},{"id":"5716f3a2001122dd8dd686d56780bfa25bc0c88e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-24 05:05:20.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3a0399c796d9b57abcd82d2d1e13306a46251447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-24 05:11:50.000000000","message":"Patch Set 4: Verified-1\n\n(46 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67bb9b2f1c78490a99df221f6a519a50\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcdc24c8532a47018397ab3a9882f13b : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b525ff9021b84f4e98b2b9b3ba828318 : FAILURE in 2m 45s","accounts_in_message":[],"_revision_number":4},{"id":"be7f85aa463382c0b6d5deeac2ec305059900217","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-24 05:17:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"292309a04355842e28a735ea1824acbe99485881","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-24 05:18:42.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"81545cfc988a3587fe268fb6860cb3927a1e9f32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-24 05:25:27.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dea304039a4f45ca8b6362ef91bfe36c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/647da836e7054306b2a97bbd3c975955 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b3c47fad874084be6177ad31eba02a : FAILURE in 2m 31s","accounts_in_message":[],"_revision_number":6},{"id":"81a588ec166dca33ccf9b359353297e287939bef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-24 05:38:51.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"7c3995d8e74653b86825978da2954f54935a159d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-24 05:46:25.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/089eb1bc1df24991beb4a4847dd84cbe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e23ecbace548999554a1def7e4f1ec : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a16d7eeee7b4100ae1e8dd5edfbf047 : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":7},{"id":"cd5ef7b1bd7425957dae0e08c21124f254309c2d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-24 05:48:49.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"30316769b38614ac73083974d639815314d735f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-24 05:50:20.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"1e05b4d1a2020d9b69fe987174dff18858599747","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-24 05:57:04.000000000","message":"Patch Set 8: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5327bd3cbb9f4f4d979db55f77eb14cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dcefb4af9044fad8415741dd6aad00c : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d1a80fe61a4d99819111063e9dc68c : FAILURE in 2m 20s","accounts_in_message":[],"_revision_number":8},{"id":"2e083f39d0a0f2bf35e144dbb898cf1300856560","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-24 06:10:08.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"9a43f1771cb91aa30d25d6360a6f1b991e22e73f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-24 06:19:13.000000000","message":"Patch Set 9: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58c5e7ba1bac40329c645bf628b41141\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62c897abae6e48c9b77378b444ce5c3d : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76840fc3aa304234b469a7f643ddd8cb : FAILURE in 2m 39s","accounts_in_message":[],"_revision_number":9},{"id":"65d4736ee8dd5c9ebd5692f1023cfc19bf5e61bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-09-24 06:21:40.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"0147baf9e95870b3ff1b30d9ab5a58963426782b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-24 06:30:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00144da7d6844e3fac3b16853527e0bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e10a558d02348eca8a621983ebb7c34 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aae850cde074c7d9b68e91bce8f6ffa : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":10},{"id":"51dfa81328cee5b8cd8b1ca392422843853c505a","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-10-27 18:45:21.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"1f7d782eadfaa3c2cbd8a2db70d1815fb6824b7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-11-04 04:34:57.000000000","message":"Patch Set 11: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"e48e1fc87772d02a52927a524683857b5c1ec84c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-11-04 04:35:25.000000000","message":"Patch Set 12: Commit message was updated.","accounts_in_message":[],"_revision_number":12},{"id":"ac014992d6132f14e7cd4491e6fe61945a41f807","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-04 04:46:08.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7d288fbd0b94f589b1c967734239db4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6193c9126d5a43858eb2cc85cdf4730d : SUCCESS in 8m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aafdd78b24ef4d68a6ab5fe210ef498f : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":12},{"id":"507847be9c475a380deffb891d6be62ee866f00d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-11-08 14:31:32.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"f71031f778895634d315d9a225ba6eb250d1ba25","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-11-08 14:32:03.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"7e0d63a67ac5d0e2365af323eff3c081bc320ad7","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-11-08 14:33:57.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"e21e9d997e7a82c5660c9fbcb74bc4805bccf62e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-08 14:44:19.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22ce8fd985304fc4902c48b65c57bd05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54347fa42d0c42ea95d59cea5398a8f7 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b211125b0d04b40a1e115c54151844b : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":13},{"id":"9cb6404fa5a0eefa79486a27492e0ef154e3fd76","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-11-10 15:46:34.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"26209424223286cc055ce1d69ba2dd2cad2369a5","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-11-10 21:07:38.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"8560f0870efa2250cfa04a9c525ab1d9ffd399a1","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-11-21 14:19:59.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"df7c7bc857d35cd6404bc64388baa57300b1eb5a","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-11-28 05:41:18.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"8deabcfe20a5c4f954190543d51abb3b586a6553","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-11-29 15:21:02.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"670d69e8a473f6bb5051321758b40ed60c449a15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 15:32:48.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc6bcea367a646dcb09ac78027a3ddcb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b92a8ae4a6914effb52d3601657eecde : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e6c6294fb2b4704b540ae6ea2031008 : FAILURE in 2m 48s","accounts_in_message":[],"_revision_number":14},{"id":"04214b231db25a9cb9224fa23253045f84deb835","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-11-30 10:57:11.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"fe98a565998f1653fe544591d5a19a836298eea0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 11:05:15.000000000","message":"Patch Set 15: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cc8fb787ca54f9eb44718c7edb1c1c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cc33bc088074be89f2517007a655e4b : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/644494e6810249a8b65c0a1a57070295 : FAILURE in 2m 54s","accounts_in_message":[],"_revision_number":15},{"id":"89bce58a1a15e196710b6b94d56af0216e3cbbd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-11-30 11:09:59.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"70f725a25c418b702f7035f2d50a0a32b9134720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 11:19:26.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0eb4c0d578334635b69c99494e83b1a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c062630c68cb4621be58681fe76d9834 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e19603e64c294598b40d48e924e429fc : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":16},{"id":"46022cd4d67f1cb63fed414ee0972bc04133ade2","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2023-12-12 08:16:05.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"3cdc073cffd32e0942e14c12de4a40da666aca7a","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2024-02-24 03:38:13.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"59bd3c7b2141394da543980fdaae07037cfacaee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2024-02-28 14:18:58.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"537d380a0c857638017be409dd1a1fe62e9540c5","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2024-02-28 14:20:04.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"23fdd892ba47f082cf525b7844dc7806d4353c4a","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-02-28 14:20:37.000000000","message":"Patch Set 17: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"61bfe53aaa9c4e608c075422a3c57baa4114c2cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 14:32:45.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14194cd1d409471ba48aef691a240140\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe46e9deb8e74aba8532922748f5b3ec : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35b1fdf269a498a8097c88cf7bd645b : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":17},{"id":"5501caf7b9840d2a84b83b506079f8ef74636d52","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-03-19 13:17:32.000000000","message":"Patch Set 17: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"1a6a55944240ab0db81bb5ce141ff5faa74898db","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2024-03-28 15:17:10.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"d8213812aac7a9b88458be4c028584b6de7aead4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2024-03-28 15:17:22.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"b83b5831ba472808bf9af3397b2dc7ab9f224d09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 15:29:27.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3e4b4f2c9414d5eab002745eba2dabb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/508bfcc64e6e48b0b68a4b5a81b917e6 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61641329f1ce4c83b4394eaccaf4649b : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":18},{"id":"9a29a4e2bf18730d6e90f21382811887475aa17d","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-04-02 21:10:55.000000000","message":"Patch Set 18: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"52e45cbd7985aa0d09243d5a84cc433f4cf2b65c","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-04-02 21:19:56.000000000","message":"Patch Set 18:\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"609410288a3517c20ff21b8dce9017e43393a09f","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-04-02 21:20:00.000000000","message":"Patch Set 18: -Code-Review","accounts_in_message":[],"_revision_number":18},{"id":"49b0c3cd351f5d25ae6e7891cdc2ae81b02eede7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2024-04-20 04:03:52.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"fd7b61e6f84042f8d5c43efa93687fbec531df1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 04:15:17.000000000","message":"Patch Set 19: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5d67268a9bd46089e5d6e59a6d23157\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b83f83175eb24ce0a7ea6bf9ddf12ef8 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d8205a897a54625a8595915c73ff2c9 : FAILURE in 2m 35s","accounts_in_message":[],"_revision_number":19},{"id":"fa189049080365081e23c45a157e78d079299aba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2024-04-20 04:22:51.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"a11bc6bbd856b6a34bf0ebd6bf2a0e1ea0b96949","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2024-04-20 04:24:27.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"81d3197a8ff1ff3904ac7cba1b2776d2587404bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-20 04:29:02.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f46563a3f3342d1ae4d4d7d202ef5fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/769595def7e14124befe02eaf4491422 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a06e051e6d8f43b3bfc3bdb3cc8516b2 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":20},{"id":"b7e74f9610d9f98e65499f851017209251ac4054","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-05-08 13:29:03.000000000","message":"Patch Set 20: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"a7b88f09a26f0b94caa9b2b79d0dac23da0d4f26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2024-05-08 14:04:53.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"9728496f5088dc9ad404769ac7d9d49e653d7af9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 14:43:00.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44f4c6ee1d784e6aaad7287d0f10b2bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/280a2f140ac44c7d917850a8b95d83c5 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01dc4ee86e8e4ff685009196fbda6571 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":21},{"id":"2bc650d46086e574f390f90280e421166c90c5cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2024-05-21 13:18:03.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"9a13c8982338998eb5360a191715e3215199f39d","author":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"date":"2024-05-21 13:22:28.000000000","message":"Patch Set 22:\n\n(4 comments)","accounts_in_message":[],"_revision_number":22},{"id":"b03c01e42cc02c3a6ceb317b4f6517c3e9e2d17d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 13:26:28.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dddbc2167fe243c3b61b5c21ddeb0c05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e61b582c1e14188ab13a89bc24d4988 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a304aedf37144fbaeb658d44eea6509 : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":22},{"id":"d307a75653fbce69cab1483c1fee15905b66ef5f","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2024-05-22 14:00:28.000000000","message":"Patch Set 22: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"2b1b7550a6dbd08d32a858f890c2bbf2c8d75fc1","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2024-05-22 14:07:06.000000000","message":"Patch Set 22: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"5530b5bc8cabbf48f31072da751d04781ad2cc04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 14:07:25.000000000","message":"Patch Set 22: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":22},{"id":"c46c4b39e435a8038e37212c29dd35ace1ec54d0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-05-22 14:09:47.000000000","message":"Patch Set 22:\n\n(2 comments)","accounts_in_message":[],"_revision_number":22},{"id":"3e3622c9729cd00efe34a5d9e1669db397065049","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 14:15:47.000000000","message":"Patch Set 22: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e21b491eeb3c49b98cc85ce89e02a89e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17cc8cf5530749af86931e9f0746d766 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d80212a7c171460cbc3fac8895d70304 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":22},{"id":"c71d96afc5bf71244dbc49112f5bddd574224abf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 14:15:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":22},{"id":"d3d154e6787629c63aef92fa41059122365d4eb3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 14:17:05.000000000","message":"Patch Set 22:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5e30004e9da4d9382bc023d5022ef46\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11aec70b19c44f5887e98ac1c6b0f4d0 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":22}],"current_revision_number":22,"current_revision":"ea37664d79a9bc3c089f176bb098d476210bb83e","revisions":{"04560cfe8bf13c343f8faf406d1c30a408257f07":{"kind":"REWORK","_number":1,"created":"2023-09-23 09:09:17.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/1"}}},"commit":{"parents":[{"commit":"66888cbd31ef4013e3a958648af233bd81e4f96e","subject":"Fix doc-pot-filter.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/66888cbd31ef4013e3a958648af233bd81e4f96e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nstory: 2010915\ntast: 48768\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/04560cfe8bf13c343f8faf406d1c30a408257f07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/04560cfe8bf13c343f8faf406d1c30a408257f07"}]},"branch":"refs/heads/master"},"462c329dd243587fab76c9bf0157dcbf2fa17bf9":{"kind":"REWORK","_number":2,"created":"2023-09-24 04:41:51.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/2"}}},"commit":{"parents":[{"commit":"66888cbd31ef4013e3a958648af233bd81e4f96e","subject":"Fix doc-pot-filter.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/66888cbd31ef4013e3a958648af233bd81e4f96e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-24 04:41:37.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nstory: 2010915\ntast: 48768\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/462c329dd243587fab76c9bf0157dcbf2fa17bf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/462c329dd243587fab76c9bf0157dcbf2fa17bf9"}]},"branch":"refs/heads/master"},"1dea7129d0b9c70d2589f615fb136f306a20fa73":{"kind":"REWORK","_number":3,"created":"2023-09-24 04:44:16.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/3"}}},"commit":{"parents":[{"commit":"66888cbd31ef4013e3a958648af233bd81e4f96e","subject":"Fix doc-pot-filter.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/66888cbd31ef4013e3a958648af233bd81e4f96e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-24 04:44:05.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nstory: 2010915\ntast: 48768\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/1dea7129d0b9c70d2589f615fb136f306a20fa73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/1dea7129d0b9c70d2589f615fb136f306a20fa73"}]},"branch":"refs/heads/master"},"f73794d53453bc72bc501926e8180f0654741c1f":{"kind":"REWORK","_number":4,"created":"2023-09-24 05:05:20.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/4"}}},"commit":{"parents":[{"commit":"66888cbd31ef4013e3a958648af233bd81e4f96e","subject":"Fix doc-pot-filter.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/66888cbd31ef4013e3a958648af233bd81e4f96e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-24 05:05:09.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nstory: 2010915\ntast: 48768\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/f73794d53453bc72bc501926e8180f0654741c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/f73794d53453bc72bc501926e8180f0654741c1f"}]},"branch":"refs/heads/master"},"da63d01b52b7ab0dfac15d980d5e598e17eba303":{"kind":"REWORK","_number":5,"created":"2023-09-24 05:17:14.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/5"}}},"commit":{"parents":[{"commit":"66888cbd31ef4013e3a958648af233bd81e4f96e","subject":"Fix doc-pot-filter.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/66888cbd31ef4013e3a958648af233bd81e4f96e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-24 05:17:05.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nstory: 2010915\ntast: 48768\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/da63d01b52b7ab0dfac15d980d5e598e17eba303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/da63d01b52b7ab0dfac15d980d5e598e17eba303"}]},"branch":"refs/heads/master"},"9a8211ed0eccfde90cf63c61557aeaae6e581836":{"kind":"REWORK","_number":6,"created":"2023-09-24 05:18:42.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/6"}}},"commit":{"parents":[{"commit":"66888cbd31ef4013e3a958648af233bd81e4f96e","subject":"Fix doc-pot-filter.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/66888cbd31ef4013e3a958648af233bd81e4f96e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-24 05:18:33.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nstory: 2010915\ntast: 48768\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/9a8211ed0eccfde90cf63c61557aeaae6e581836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/9a8211ed0eccfde90cf63c61557aeaae6e581836"}]},"branch":"refs/heads/master"},"5d693c96d1ee3b7405253d0817d249fddc13a8de":{"kind":"REWORK","_number":7,"created":"2023-09-24 05:38:51.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/7"}}},"commit":{"parents":[{"commit":"66888cbd31ef4013e3a958648af233bd81e4f96e","subject":"Fix doc-pot-filter.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/66888cbd31ef4013e3a958648af233bd81e4f96e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-24 05:38:40.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nstory: 2010915\ntast: 48768\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/5d693c96d1ee3b7405253d0817d249fddc13a8de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/5d693c96d1ee3b7405253d0817d249fddc13a8de"}]},"branch":"refs/heads/master"},"435d41c94d467e0a93ef70a5266295604cb1765b":{"kind":"REWORK","_number":8,"created":"2023-09-24 05:50:20.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/8"}}},"commit":{"parents":[{"commit":"66888cbd31ef4013e3a958648af233bd81e4f96e","subject":"Fix doc-pot-filter.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/66888cbd31ef4013e3a958648af233bd81e4f96e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-24 05:50:10.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nstory: 2010915\ntast: 48768\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/435d41c94d467e0a93ef70a5266295604cb1765b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/435d41c94d467e0a93ef70a5266295604cb1765b"}]},"branch":"refs/heads/master"},"2f9c908b3c6d597bb334cd4503b94e06efa29a8d":{"kind":"REWORK","_number":9,"created":"2023-09-24 06:10:08.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/9"}}},"commit":{"parents":[{"commit":"66888cbd31ef4013e3a958648af233bd81e4f96e","subject":"Fix doc-pot-filter.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/66888cbd31ef4013e3a958648af233bd81e4f96e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-24 06:09:58.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nstory: 2010915\ntast: 48768\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/2f9c908b3c6d597bb334cd4503b94e06efa29a8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/2f9c908b3c6d597bb334cd4503b94e06efa29a8d"}]},"branch":"refs/heads/master"},"fb84a414da9877051a08073adcfe53de9da31278":{"kind":"REWORK","_number":10,"created":"2023-09-24 06:21:40.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/10"}}},"commit":{"parents":[{"commit":"66888cbd31ef4013e3a958648af233bd81e4f96e","subject":"Fix doc-pot-filter.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/66888cbd31ef4013e3a958648af233bd81e4f96e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-24 06:21:04.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nstory: 2010915\ntast: 48768\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/fb84a414da9877051a08073adcfe53de9da31278"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/fb84a414da9877051a08073adcfe53de9da31278"}]},"branch":"refs/heads/master"},"3812782be46659c609110286cfd1130e028ffb8a":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2023-11-04 04:34:57.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/11"}}},"commit":{"parents":[{"commit":"66888cbd31ef4013e3a958648af233bd81e4f96e","subject":"Fix doc-pot-filter.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/66888cbd31ef4013e3a958648af233bd81e4f96e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong Jeong","email":"gsniper777@gmail.com","date":"2023-11-04 04:34:57.000000000","tz":0},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nImplements: [blueprint weblate-tools](https://blueprints.launchpad.net/openstack/?searchtext\u003dweblate-tools)\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/3812782be46659c609110286cfd1130e028ffb8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/3812782be46659c609110286cfd1130e028ffb8a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"21fce7090833887e7cfba335a73c8c10552789b1":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2023-11-04 04:35:25.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/12"}}},"commit":{"parents":[{"commit":"66888cbd31ef4013e3a958648af233bd81e4f96e","subject":"Fix doc-pot-filter.sh","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/66888cbd31ef4013e3a958648af233bd81e4f96e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong Jeong","email":"gsniper777@gmail.com","date":"2023-11-04 04:35:25.000000000","tz":0},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nImplements: blueprint weblate-tools\n - https://blueprints.launchpad.net/openstack/?searchtext\u003dweblate-tools\n\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/21fce7090833887e7cfba335a73c8c10552789b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/21fce7090833887e7cfba335a73c8c10552789b1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"15e1a8b0369de23492c2b998d8b71d76849ef557":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-11-08 14:31:32.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/13"}}},"commit":{"parents":[{"commit":"d5d64997abe918b76ff03c44d402ed14d5e18d28","subject":"Merge \"generatepot.sh: Drop UUID filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/d5d64997abe918b76ff03c44d402ed14d5e18d28"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong Jeong","email":"gsniper777@gmail.com","date":"2023-11-08 14:31:32.000000000","tz":0},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nImplements: blueprint weblate-tools\n - https://blueprints.launchpad.net/openstack/?searchtext\u003dweblate-tools\n\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/15e1a8b0369de23492c2b998d8b71d76849ef557"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/15e1a8b0369de23492c2b998d8b71d76849ef557"}]},"branch":"refs/heads/master","description":"Rebase"},"74501f4ced281cb7905413175c3007bbdd905a49":{"kind":"REWORK","_number":14,"created":"2023-11-29 15:21:02.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/14"}}},"commit":{"parents":[{"commit":"d5d64997abe918b76ff03c44d402ed14d5e18d28","subject":"Merge \"generatepot.sh: Drop UUID filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/d5d64997abe918b76ff03c44d402ed14d5e18d28"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Floodnut","email":"gsniper777@gmail.com","date":"2023-11-29 15:20:59.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nImplements: blueprint weblate-tools\n - https://blueprints.launchpad.net/openstack/?searchtext\u003dweblate-tools\n\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/74501f4ced281cb7905413175c3007bbdd905a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/74501f4ced281cb7905413175c3007bbdd905a49"}]},"branch":"refs/heads/master"},"86e384cc02d9910ff314280978316d5d05c2cf11":{"kind":"REWORK","_number":15,"created":"2023-11-30 10:57:11.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/15"}}},"commit":{"parents":[{"commit":"d5d64997abe918b76ff03c44d402ed14d5e18d28","subject":"Merge \"generatepot.sh: Drop UUID filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/d5d64997abe918b76ff03c44d402ed14d5e18d28"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Floodnut","email":"gsniper777@gmail.com","date":"2023-11-30 10:56:55.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nImplements: blueprint weblate-tools\n - https://blueprints.launchpad.net/openstack/?searchtext\u003dweblate-tools\n\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/86e384cc02d9910ff314280978316d5d05c2cf11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/86e384cc02d9910ff314280978316d5d05c2cf11"}]},"branch":"refs/heads/master"},"ae7cbb9bd0023c3bcaa1107e52e73ebf9a16c70a":{"kind":"REWORK","_number":16,"created":"2023-11-30 11:09:59.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/16"}}},"commit":{"parents":[{"commit":"d5d64997abe918b76ff03c44d402ed14d5e18d28","subject":"Merge \"generatepot.sh: Drop UUID filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/d5d64997abe918b76ff03c44d402ed14d5e18d28"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Floodnut","email":"gsniper777@gmail.com","date":"2023-11-30 11:09:42.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nImplements: blueprint weblate-tools\n - https://blueprints.launchpad.net/openstack/?searchtext\u003dweblate-tools\n\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/ae7cbb9bd0023c3bcaa1107e52e73ebf9a16c70a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/ae7cbb9bd0023c3bcaa1107e52e73ebf9a16c70a"}]},"branch":"refs/heads/master"},"d3a062cd63080f52bc993ade659fa7ee542d3f6b":{"kind":"REWORK","_number":17,"created":"2024-02-28 14:18:58.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/17"}}},"commit":{"parents":[{"commit":"d5d64997abe918b76ff03c44d402ed14d5e18d28","subject":"Merge \"generatepot.sh: Drop UUID filtering\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/d5d64997abe918b76ff03c44d402ed14d5e18d28"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong Jeong","email":"gsniper777@gmail.com","date":"2024-02-28 14:18:57.000000000","tz":0},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nImplements: blueprint weblate-tools\n - https://blueprints.launchpad.net/openstack/?searchtext\u003dweblate-tools\n\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/d3a062cd63080f52bc993ade659fa7ee542d3f6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/d3a062cd63080f52bc993ade659fa7ee542d3f6b"}]},"branch":"refs/heads/master"},"95c6a535f3f9cd7ed717f9bbb34b8376e676198a":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-03-28 15:17:22.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/18"}}},"commit":{"parents":[{"commit":"84b571f65c1f41b796933aec09c940dc64b9bdc9","subject":"Merge \"Add weblate users\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/84b571f65c1f41b796933aec09c940dc64b9bdc9"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong Jeong","email":"gsniper777@gmail.com","date":"2024-03-28 15:17:22.000000000","tz":0},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nImplements: blueprint weblate-tools\n - https://blueprints.launchpad.net/openstack/?searchtext\u003dweblate-tools\n\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/95c6a535f3f9cd7ed717f9bbb34b8376e676198a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/95c6a535f3f9cd7ed717f9bbb34b8376e676198a"}]},"branch":"refs/heads/master","description":"Rebase"},"0197cfb1a8ee7ce65cfb7bf7c730c893879e7dff":{"kind":"REWORK","_number":19,"created":"2024-04-20 04:03:52.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/19"}}},"commit":{"parents":[{"commit":"90de0f20d307f27f7a2185a864a6fe55b469818e","subject":"Merge \"[doc] IRC Meeting information update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/90de0f20d307f27f7a2185a864a6fe55b469818e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2024-04-20 04:03:33.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nImplements: blueprint weblate-tools\n - https://blueprints.launchpad.net/openstack/?searchtext\u003dweblate-tools\n\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/0197cfb1a8ee7ce65cfb7bf7c730c893879e7dff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/0197cfb1a8ee7ce65cfb7bf7c730c893879e7dff"}]},"branch":"refs/heads/master"},"5ac34cca40d615b9be575c39ef554b9a5ed6a87e":{"kind":"REWORK","_number":20,"created":"2024-04-20 04:22:51.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/20"}}},"commit":{"parents":[{"commit":"90de0f20d307f27f7a2185a864a6fe55b469818e","subject":"Merge \"[doc] IRC Meeting information update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/90de0f20d307f27f7a2185a864a6fe55b469818e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2024-04-20 04:22:38.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nImplements: blueprint weblate-tools\n - https://blueprints.launchpad.net/openstack/?searchtext\u003dweblate-tools\n\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/5ac34cca40d615b9be575c39ef554b9a5ed6a87e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/5ac34cca40d615b9be575c39ef554b9a5ed6a87e"}]},"branch":"refs/heads/master"},"22dfe38af5128121fbdf6a95328482c1149bd76f":{"kind":"REWORK","_number":21,"created":"2024-05-08 14:04:53.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/21"}}},"commit":{"parents":[{"commit":"90de0f20d307f27f7a2185a864a6fe55b469818e","subject":"Merge \"[doc] IRC Meeting information update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/90de0f20d307f27f7a2185a864a6fe55b469818e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2024-05-08 14:04:30.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nImplements: blueprint weblate-tools\n - https://blueprints.launchpad.net/openstack/?searchtext\u003dweblate-tools\n\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/22dfe38af5128121fbdf6a95328482c1149bd76f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/22dfe38af5128121fbdf6a95328482c1149bd76f"}]},"branch":"refs/heads/master"},"ea37664d79a9bc3c089f176bb098d476210bb83e":{"kind":"REWORK","_number":22,"created":"2024-05-21 13:18:03.000000000","uploader":{"_account_id":36162,"name":"Kuemjong Jeong","display_name":"Goldenbell","email":"gsniper777@gmail.com","username":"jjongmin77"},"ref":"refs/changes/06/896306/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/06/896306/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/06/896306/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/06/896306/22"}}},"commit":{"parents":[{"commit":"90de0f20d307f27f7a2185a864a6fe55b469818e","subject":"Merge \"[doc] IRC Meeting information update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/90de0f20d307f27f7a2185a864a6fe55b469818e"}]}],"author":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2023-09-23 09:08:53.000000000","tz":540},"committer":{"name":"Kuemjong","email":"gsniper777@gmail.com","date":"2024-05-21 13:17:10.000000000","tz":540},"subject":"migrate zanata_stats to weblate_stats","message":"migrate zanata_stats to weblate_stats\n\n- weblate_stats.py\n- weblate_records.py\n- __init__.py\n\nImplements: blueprint weblate-tools\n - https://blueprints.launchpad.net/openstack/?searchtext\u003dweblate-tools\n\nChange-Id: Ie196e43f418e730d6fee7f0a107d198e7daacbc4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/ea37664d79a9bc3c089f176bb098d476210bb83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/ea37664d79a9bc3c089f176bb098d476210bb83e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
