)]}'
{"id":"openstack%2Fi18n~896983","triplet_id":"openstack%2Fi18n~master~If2379d669332538f7518aff65d926b0f5c41be75","project":"openstack/i18n","branch":"master","topic":"update-readme","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2023-10-08 13:36:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If2379d669332538f7518aff65d926b0f5c41be75","subject":"Update readme for i18n repo","status":"MERGED","created":"2023-10-01 17:06:24.000000000","updated":"2023-10-08 13:37:49.000000000","submitted":"2023-10-08 13:36:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896983-update-readme","meta_rev_id":"380c2e27a4a271f81e7a624ed2b138f7313036f0","_number":896983,"virtual_id_number":896983,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-08 13:36:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":2,"date":"2023-10-02 01:37:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"all":[{"value":1,"date":"2023-10-08 13:29:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-01 17:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-02 01:37:11.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"}],"messages":[{"id":"5b48e12024188cfa385d8dcde54f17acd1190820","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2023-10-01 17:06:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0291072f44131c9d6268128598d4849c8ee9d21a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-01 17:12:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de35484f18984353b6f0b36a8bb8f3b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f4a5acd687740f4aa7f2a5a83e2e407 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9df9128dfa2438c910aa20df27179c1 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ac5c832f00074dc4417ac84a837b9ec4359b3282","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-10-02 01:37:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6a6e2fd14a595e3bd48e2b2baeac61887bfe6274","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2023-10-08 13:29:17.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35287d71399f6fafadbc66cfc3bc6d18f2bc6c17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 13:29:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1644019eca895505ebb186ba0eafdf4cf3377cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 13:36:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/113b4ce4b17d440488481a2cec85903c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b4ae0d387b74c8eaeaa3ac5beec64ea : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40ba7667a4d14d6eb7e3037e36181a15 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":1},{"id":"0134676ec03474976c3e5ac1069cc1a2d90fa01d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 13:36:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"380c2e27a4a271f81e7a624ed2b138f7313036f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 13:37:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/293846985eb84cc1a4faaf9fb2364193\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d1f5d05cf634c798f059ded8f7007d1 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"814a553af02f3f0c930e15bc8bd58c82181add17","revisions":{"814a553af02f3f0c930e15bc8bd58c82181add17":{"kind":"REWORK","_number":1,"created":"2023-10-01 17:06:24.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/83/896983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/83/896983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/83/896983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/83/896983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/83/896983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/83/896983/1"}}},"commit":{"parents":[{"commit":"b03da458e68a056dc7a68c6602c46614598e0862","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/b03da458e68a056dc7a68c6602c46614598e0862"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-10-01 17:05:41.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2023-10-01 17:05:41.000000000","tz":540},"subject":"Update readme for i18n repo","message":"Update readme for i18n repo\n\n- Delete team and repository tags which no longer applies\n- Remove IRC team meeting since we do not hold now\n\nChange-Id: If2379d669332538f7518aff65d926b0f5c41be75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/814a553af02f3f0c930e15bc8bd58c82181add17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/814a553af02f3f0c930e15bc8bd58c82181add17"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
