)]}'
{"id":"openstack%2Fi18n~954846","triplet_id":"openstack%2Fi18n~master~Ic1ab6afbfe55f55fa57411f544e49f7dd3451031","project":"openstack/i18n","branch":"master","topic":"team-member-update","attention_set":{},"removed_from_attention_set":{"14482":{"account":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"last_update":"2025-08-29 11:03:20.000000000","reason":"Change was submitted"},"23279":{"account":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"last_update":"2025-08-29 10:40:56.000000000","reason":"\u003cGERRIT_ACCOUNT_23279\u003e replied on the change","reason_account":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}}},"hashtags":[],"change_id":"Ic1ab6afbfe55f55fa57411f544e49f7dd3451031","subject":"Sync translation_team.yaml with Zanata","status":"MERGED","created":"2025-07-12 04:58:38.000000000","updated":"2025-08-29 11:07:33.000000000","submitted":"2025-08-29 11:03:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"954846-team-member-update","meta_rev_id":"95fc9be2477960ca72d64716d2727a6bae8b0d01","_number":954846,"virtual_id_number":954846,"owner":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-29 11:03:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-29 10:40:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-29 10:40:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-12 05:10:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-29 06:39:24.000000000","updated_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"reviewer":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"state":"REVIEWER"}],"messages":[{"id":"12763c4387dc0ebfeda1623c4a21be1b146c79ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2025-07-12 04:58:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3104d20f0f7bb0328ab0a842a0ceb080a8555d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-12 05:10:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cad237489b73418299b11b8e3e917314\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17b9000496af42a680c6df6a3358228c : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5caf8e8f8e5b4c188e955c1a0f9711b0 : SUCCESS in 2m 26s","accounts_in_message":[],"_revision_number":1},{"id":"1654eb6e2f33c7cf2b203924e8f57975263c848d","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-08-29 06:39:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e031cb312dcc2e6c4bee19aa735810a419edbc40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 06:39:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"628a51aa2777f593993a6511699a501157d0bf18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 06:46:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/951c9d3ad17943c696d7386eceaaa316\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b30c215b385749ab84f5ed86b384f790 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bed23ed5712644db8670acc06af866df : FAILURE in 2m 23s","accounts_in_message":[],"_revision_number":1},{"id":"be6e5fd13931b9b03688c0308f491705681465cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-08-29 09:48:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"472bca9aeef50b85cb4d86b3d2218e7a909fd9b4","author":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"date":"2025-08-29 10:40:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bb0190a198a8c561abe79a79951b44946ef1613f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 10:52:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/862d34db393042c083cf86604f04efd4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d37c5403ed45b8af2d9f1282b65008 : SUCCESS in 8m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb198375cba543e09a45305f3380cd7a : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":2},{"id":"daaeb5ca2581d10dc3d66946d9eeda918d2cd3ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 10:52:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d83496f5ec3b05eddcefc510353486d522ca8c47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 11:03:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5927000df7f4fabbbe6502dc78c8f27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f57e35ac783438ea7780c9699aea567 : SUCCESS in 9m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb99c893811047078d3d583b80a5d5c3 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"6e43a7b220e1b8f4c77f33314c7e855034c5ce0c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 11:03:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"95fc9be2477960ca72d64716d2727a6bae8b0d01","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-29 11:07:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/505076bb10d246018d9935feced3fdd5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13a2ef3242f84ba4b5881eb73b2e881c : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7b8768b5c0e38fbf653d1d0b03f8c08aec60d788","revisions":{"6db733a66a4fc68a76a46fdc38f3a5322a9f8946":{"kind":"REWORK","_number":1,"created":"2025-07-12 04:58:38.000000000","uploader":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"ref":"refs/changes/46/954846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/46/954846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/46/954846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/46/954846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/46/954846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/46/954846/1"}}},"commit":{"parents":[{"commit":"651aa2bc237ea98fc94cf8dae16dc417905611fe","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/651aa2bc237ea98fc94cf8dae16dc417905611fe"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-07-12 04:51:21.000000000","tz":540},"committer":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-07-12 04:58:25.000000000","tz":540},"subject":"Sync translation_team.yaml with Zanata","message":"Sync translation_team.yaml with Zanata\n\nhis list is generated by running \"tox -e zanata-users-sync\"\ncommand to sync translation members up-to-date similar as\nI312c8c5d9299db29a4bf20c61429b22a83eff97b\n\nChange-Id: Ic1ab6afbfe55f55fa57411f544e49f7dd3451031\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/6db733a66a4fc68a76a46fdc38f3a5322a9f8946"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/6db733a66a4fc68a76a46fdc38f3a5322a9f8946"}]},"branch":"refs/heads/master"},"7b8768b5c0e38fbf653d1d0b03f8c08aec60d788":{"kind":"REWORK","_number":2,"created":"2025-08-29 09:48:27.000000000","uploader":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"},"ref":"refs/changes/46/954846/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/i18n","ref":"refs/changes/46/954846/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/i18n refs/changes/46/954846/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/i18n refs/changes/46/954846/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/i18n refs/changes/46/954846/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/i18n refs/changes/46/954846/2"}}},"commit":{"parents":[{"commit":"651aa2bc237ea98fc94cf8dae16dc417905611fe","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/651aa2bc237ea98fc94cf8dae16dc417905611fe"}]}],"author":{"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","date":"2025-07-12 04:51:21.000000000","tz":540},"committer":{"name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","date":"2025-08-29 09:48:19.000000000","tz":540},"subject":"Sync translation_team.yaml with Zanata","message":"Sync translation_team.yaml with Zanata\n\nhis list is generated by running \"tox -e zanata-users-sync\"\ncommand to sync translation members up-to-date similar as\nI312c8c5d9299db29a4bf20c61429b22a83eff97b\n\nChange-Id: Ic1ab6afbfe55f55fa57411f544e49f7dd3451031\nSigned-off-by: Ian Y. Choi \u003cianyrchoi@gmail.com\u003e\nSigned-off-by: Seongsoo Cho \u003cppiyakk2@printf.kr\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/7b8768b5c0e38fbf653d1d0b03f8c08aec60d788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/i18n/commit/7b8768b5c0e38fbf653d1d0b03f8c08aec60d788"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23279,"name":"Seongsoo Cho","display_name":"Seongsoo Cho","email":"ppiyakk2@printf.kr","username":"seongsoo.cho"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
