)]}'
{"id":"openstack%2Finstack-undercloud~234216","triplet_id":"openstack%2Finstack-undercloud~master~Ief0cef5c022a8904582eb3c71a537d3cae75fde6","project":"openstack/instack-undercloud","branch":"master","topic":"enable-tempest","hashtags":[],"change_id":"Ief0cef5c022a8904582eb3c71a537d3cae75fde6","subject":"Mostly revert \"Disable tempest\"","status":"MERGED","created":"2015-10-13 13:36:45.000000000","updated":"2015-11-04 17:02:34.000000000","submitted":"2015-11-04 17:02:34.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70d05734097cfe2cf42a37e0cd0ea8876533278e","_number":234216,"virtual_id_number":234216,"owner":{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":2,"date":"2015-11-04 17:02:34.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"date":"2015-10-16 07:27:57.000000000","_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},{"value":0,"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2015-11-04 17:00:57.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2015-10-22 09:18:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2015-10-15 19:26:49.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":1,"date":"2015-10-22 07:50:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},{"value":0,"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2015-11-04 17:00:57.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},{"value":0,"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2015-10-13 14:00:26.000000000","updated_by":{"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},"reviewer":{"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},"state":"REVIEWER"},{"updated":"2015-10-15 19:26:49.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2015-10-22 07:50:39.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2015-10-22 09:18:13.000000000","updated_by":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2015-11-04 17:00:57.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2015-11-04 17:02:34.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"4e0e82e663be18268b6c34933b214a433a6cf823","author":{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},"date":"2015-10-13 13:36:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a62446133cb2075d44aa4dfb92082e18cd2a39e8","author":{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},"date":"2015-10-13 13:37:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9fa758c627c865d4302626dc159e4c871b749860","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2015-10-13 13:46:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"263ed6aefc1bababf9f0bcc8e247909aab3dae34","author":{"_account_id":12715,"name":"John Trowbridge","email":"trown@redhat.com","username":"trown"},"date":"2015-10-13 14:00:26.000000000","message":"Patch Set 2: Code-Review+1\n\nConfirmed tempest is in delorean and works.","accounts_in_message":[],"_revision_number":2},{"id":"596bb9db94e376501fe094830fd8ce342bb4f3d3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-13 15:11:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ironic-overcloud-f21puppet-ceph http://logs.openstack.org/16/234216/2/check-tripleo/gate-tripleo-ironic-overcloud-f21puppet-ceph/97578ef/ : SUCCESS in 1h 33m 41s (non-voting)\n- gate-tripleo-ironic-overcloud-f21puppet-ha http://logs.openstack.org/16/234216/2/check-tripleo/gate-tripleo-ironic-overcloud-f21puppet-ha/5a22af1/ : SUCCESS in 1h 31m 29s (non-voting)\n- gate-tripleo-ironic-overcloud-f21puppet-nonha http://logs.openstack.org/16/234216/2/check-tripleo/gate-tripleo-ironic-overcloud-f21puppet-nonha/525f2d6/ : SUCCESS in 1h 30m 39s","accounts_in_message":[],"_revision_number":2},{"id":"789f5af44b306babb9ba43889be549701f58ecf2","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2015-10-13 16:48:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"140d9b333bdb547d405282509a6bfa32d1674b1f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-13 16:57:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-pep8 http://logs.openstack.org/16/234216/2/check/gate-instack-undercloud-pep8/6854d1a/ : SUCCESS in 48s\n- gate-instack-undercloud-docs http://docs-draft.openstack.org/16/234216/2/check/gate-instack-undercloud-docs/7398f0e//doc/build/html/ : SUCCESS in 58s\n- gate-instack-undercloud-python27 http://logs.openstack.org/16/234216/2/check/gate-instack-undercloud-python27/8705f54/ : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2},{"id":"cf59d8841b838e35bbac945241dc11e8f8f6e01a","author":{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},"date":"2015-10-14 09:03:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"77c70cef4da914a619ce5bdbb891e42c9236d58b","author":{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},"date":"2015-10-14 09:03:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f1596182de857d9791a06380e9a32719a5867dda","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-14 09:06:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-pep8 http://logs.openstack.org/16/234216/3/check/gate-instack-undercloud-pep8/efae1bb/ : SUCCESS in 44s\n- gate-instack-undercloud-docs http://docs-draft.openstack.org/16/234216/3/check/gate-instack-undercloud-docs/c701d72//doc/build/html/ : SUCCESS in 1m 02s\n- gate-instack-undercloud-python27 http://logs.openstack.org/16/234216/3/check/gate-instack-undercloud-python27/6acda3f/ : SUCCESS in 52s","accounts_in_message":[],"_revision_number":3},{"id":"ab0689bab101244d8ff12574dfc9b7fc6276986f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-10-14 10:50:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ironic-overcloud-f21puppet-ceph http://logs.openstack.org/16/234216/3/check-tripleo/gate-tripleo-ironic-overcloud-f21puppet-ceph/888cb1e/ : SUCCESS in 1h 47m 04s (non-voting)\n- gate-tripleo-ironic-overcloud-f21puppet-ha http://logs.openstack.org/16/234216/3/check-tripleo/gate-tripleo-ironic-overcloud-f21puppet-ha/64623c4/ : SUCCESS in 1h 43m 04s (non-voting)\n- gate-tripleo-ironic-overcloud-f21puppet-nonha http://logs.openstack.org/16/234216/3/check-tripleo/gate-tripleo-ironic-overcloud-f21puppet-nonha/c0cb42f/ : SUCCESS in 1h 39m 25s","accounts_in_message":[],"_revision_number":3},{"id":"c5829744c79f5af293211f332992a18061e0d2d7","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2015-10-15 19:26:49.000000000","message":"Patch Set 3: Code-Review+2\n\nMostly +2 :)","accounts_in_message":[],"_revision_number":3},{"id":"ed796238428c20a9b93dc84ab7117afe495e84ff","author":{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},"date":"2015-10-16 07:27:57.000000000","message":"Patch Set 3:\n\nThanks James! :D","accounts_in_message":[],"_revision_number":3},{"id":"9d5dca33d2c5eaa57aa62c7236c215da94d53d99","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2015-10-22 07:50:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0c2f18f72dee1a204c1427de12139a599048f955","author":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"date":"2015-10-22 09:18:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"64f319b0c509cee3e12f9c32dec1d0178cbc8205","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2015-11-04 17:00:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nMostly approved","accounts_in_message":[],"_revision_number":3},{"id":"4a57c0c620d0cfcf6706998d277095aab39db03e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-04 17:01:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"604ddb9f1a213048e8420e090dae70b187be09fc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-04 17:02:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-instack-undercloud-docs http://docs-draft.openstack.org/16/234216/3/gate/gate-instack-undercloud-docs/4a11d35//doc/build/html/ : SUCCESS in 52s\n- gate-instack-undercloud-pep8 http://logs.openstack.org/16/234216/3/gate/gate-instack-undercloud-pep8/0efbb39/ : SUCCESS in 44s\n- gate-instack-undercloud-python27 http://logs.openstack.org/16/234216/3/gate/gate-instack-undercloud-python27/b80e473/ : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3},{"id":"5cda9a024a84998723a515826c79de9005e79dd9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2015-11-04 17:02:34.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5e0d8542bb42e2cbc349d5ac0d1b07c765633e01","revisions":{"73cac8130ba7f61c7a46f3ca005306694a6c79e9":{"kind":"REWORK","_number":1,"created":"2015-10-13 13:36:45.000000000","uploader":{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},"ref":"refs/changes/16/234216/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/16/234216/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/16/234216/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/16/234216/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/16/234216/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/16/234216/1"}}},"commit":{"parents":[{"commit":"f5c0daf5b9137d9516a03aa4ae0748c5e25c9df7","subject":"Merge \"Update README to include tags recognized by release-tools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/f5c0daf5b9137d9516a03aa4ae0748c5e25c9df7"}]}],"author":{"name":"Imre Farkas","email":"ifarkas@redhat.com","date":"2015-10-13 13:34:51.000000000","tz":120},"committer":{"name":"Imre Farkas","email":"ifarkas@redhat.com","date":"2015-10-13 13:36:42.000000000","tz":120},"subject":"Revert \"Disable tempest\"","message":"Revert \"Disable tempest\"\n\nThis reverts commit 78f7de73973d3676185348e3e22963096f1536d1.\nTempest is now available delorean.\n\nChange-Id: Ief0cef5c022a8904582eb3c71a537d3cae75fde6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/73cac8130ba7f61c7a46f3ca005306694a6c79e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/73cac8130ba7f61c7a46f3ca005306694a6c79e9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f5c0daf5b9137d9516a03aa4ae0748c5e25c9df7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"31a90fed178cdf3407c60fc96bb53e6d91b995d5":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2015-10-13 13:37:45.000000000","uploader":{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},"ref":"refs/changes/16/234216/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/16/234216/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/16/234216/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/16/234216/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/16/234216/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/16/234216/2"}}},"commit":{"parents":[{"commit":"f5c0daf5b9137d9516a03aa4ae0748c5e25c9df7","subject":"Merge \"Update README to include tags recognized by release-tools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/f5c0daf5b9137d9516a03aa4ae0748c5e25c9df7"}]}],"author":{"name":"Imre Farkas","email":"ifarkas@redhat.com","date":"2015-10-13 13:34:51.000000000","tz":120},"committer":{"name":"Imre Farkas","email":"ifarkas@redhat.com","date":"2015-10-13 13:37:33.000000000","tz":120},"subject":"Revert \"Disable tempest\"","message":"Revert \"Disable tempest\"\n\nThis reverts commit 78f7de73973d3676185348e3e22963096f1536d1.\nTempest is now available in delorean.\n\nChange-Id: Ief0cef5c022a8904582eb3c71a537d3cae75fde6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/31a90fed178cdf3407c60fc96bb53e6d91b995d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/31a90fed178cdf3407c60fc96bb53e6d91b995d5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f5c0daf5b9137d9516a03aa4ae0748c5e25c9df7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"5e0d8542bb42e2cbc349d5ac0d1b07c765633e01":{"kind":"REWORK","_number":3,"created":"2015-10-14 09:03:40.000000000","uploader":{"_account_id":7419,"name":"Imre Farkas","email":"farkas.imre@gmail.com","username":"ifarkas"},"ref":"refs/changes/16/234216/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/16/234216/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/16/234216/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/16/234216/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/16/234216/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/16/234216/3"}}},"commit":{"parents":[{"commit":"f5c0daf5b9137d9516a03aa4ae0748c5e25c9df7","subject":"Merge \"Update README to include tags recognized by release-tools\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/f5c0daf5b9137d9516a03aa4ae0748c5e25c9df7"}]}],"author":{"name":"Imre Farkas","email":"ifarkas@redhat.com","date":"2015-10-13 13:34:51.000000000","tz":120},"committer":{"name":"Imre Farkas","email":"ifarkas@redhat.com","date":"2015-10-14 08:49:26.000000000","tz":120},"subject":"Mostly revert \"Disable tempest\"","message":"Mostly revert \"Disable tempest\"\n\nThis reverts commit 78f7de73973d3676185348e3e22963096f1536d1 except for\nthe deprecation notes in undercloud.conf.sample.\nTempest is now available in delorean.\n\nChange-Id: Ief0cef5c022a8904582eb3c71a537d3cae75fde6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/5e0d8542bb42e2cbc349d5ac0d1b07c765633e01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/5e0d8542bb42e2cbc349d5ac0d1b07c765633e01"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"f5c0daf5b9137d9516a03aa4ae0748c5e25c9df7","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
