)]}'
{"id":"openstack%2Finstack-undercloud~291142","triplet_id":"openstack%2Finstack-undercloud~stable%2Fliberty~I40dd4543580332a46cde96f4bca639eb777ac7a8","project":"openstack/instack-undercloud","branch":"stable/liberty","topic":"bug/1358237","hashtags":[],"change_id":"I40dd4543580332a46cde96f4bca639eb777ac7a8","subject":"Use service tenant for ceilometer","status":"MERGED","created":"2016-03-10 12:25:53.000000000","updated":"2016-03-10 18:44:16.000000000","submitted":"2016-03-10 18:44:16.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9ecb2f0f0b0b8ed57595e8be681365e3317c960a","_number":291142,"virtual_id_number":291142,"owner":{"_account_id":9979,"name":"Mike Burns","email":"mburns@redhat.com","username":"mburns"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-03-10 18:44:16.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-03-10 18:35:52.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-03-10 18:35:52.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-03-10 18:35:52.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2016-03-10 18:44:16.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"3f542a23cfada7eee3cd78ef305dfd123b216bb6","author":{"_account_id":9979,"name":"Mike Burns","email":"mburns@redhat.com","username":"mburns"},"date":"2016-03-10 12:25:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"090cfe044b59f26ea48c2874d7e2d329cccbbc68","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-10 12:37:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-instack-undercloud-pep8 http://logs.openstack.org/42/291142/1/check/gate-instack-undercloud-pep8/6946a62/ : SUCCESS in 2m 58s\n- gate-instack-undercloud-docs http://docs-draft.openstack.org/42/291142/1/check/gate-instack-undercloud-docs/26a6b7c//doc/build/html/ : SUCCESS in 1m 24s\n- gate-instack-undercloud-python27 http://logs.openstack.org/42/291142/1/check/gate-instack-undercloud-python27/721a41c/ : SUCCESS in 1m 29s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/42/291142/1/check/gate-instack-undercloud-puppet-lint/d62a661/ : FAILURE in 1m 26s\n- gate-instack-undercloud-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/42/291142/1/check/gate-instack-undercloud-puppet-syntax-3-dsvm-centos7/cfab691/ : SUCCESS in 54s\n- gate-instack-undercloud-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/42/291142/1/check/gate-instack-undercloud-puppet-syntax-4-dsvm-centos7/2b4b05b/ : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1},{"id":"11f62723e930bfff2ba617d77d15ed100ac3e778","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-10 14:36:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-f22-ceph http://logs.openstack.org/42/291142/1/check-tripleo/gate-tripleo-ci-f22-ceph/ce5f44d/ : SUCCESS in 1h 46m 52s\n- gate-tripleo-ci-f22-ha http://logs.openstack.org/42/291142/1/check-tripleo/gate-tripleo-ci-f22-ha/0b4f6ad/ : SUCCESS in 1h 49m 47s\n- gate-tripleo-ci-f22-nonha http://logs.openstack.org/42/291142/1/check-tripleo/gate-tripleo-ci-f22-nonha/f58149f/ : SUCCESS in 1h 55m 15s\n- gate-tripleo-ci-f22-containers http://logs.openstack.org/42/291142/1/check-tripleo/gate-tripleo-ci-f22-containers/472bcc9/ : FAILURE in 7m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5edb8ce87e273e8156954cad0d36d977e8f6caad","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2016-03-10 18:35:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"860f5a90c29899c74874bc0798f41c1cb05035ae","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-10 18:41:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-pep8 http://logs.openstack.org/42/291142/1/check/gate-instack-undercloud-pep8/5d82ee7/ : SUCCESS in 2m 56s\n- gate-instack-undercloud-docs http://docs-draft.openstack.org/42/291142/1/check/gate-instack-undercloud-docs/92d7a30//doc/build/html/ : SUCCESS in 3m 26s\n- gate-instack-undercloud-python27 http://logs.openstack.org/42/291142/1/check/gate-instack-undercloud-python27/b66a35f/ : SUCCESS in 3m 45s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/42/291142/1/check/gate-instack-undercloud-puppet-lint/6d34572/ : SUCCESS in 3m 13s\n- gate-instack-undercloud-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/42/291142/1/check/gate-instack-undercloud-puppet-syntax-3-dsvm-centos7/7f47b4a/ : SUCCESS in 1m 16s\n- gate-instack-undercloud-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/42/291142/1/check/gate-instack-undercloud-puppet-syntax-4-dsvm-centos7/0f0b886/ : SUCCESS in 30s","accounts_in_message":[],"_revision_number":1},{"id":"e1d36d8841d2d373ebe62f797b8427671e8c3bcb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-10 18:41:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"7625e8d3157218d6d05918c55c96c95cdf245d65","date":"2016-03-10 18:44:16.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"5eda4e3a79590134df8fb5239df3cb955e130afa","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-03-10 18:44:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-instack-undercloud-docs http://docs-draft.openstack.org/42/291142/1/gate/gate-instack-undercloud-docs/8e6c903//doc/build/html/ : SUCCESS in 2m 04s\n- gate-instack-undercloud-pep8 http://logs.openstack.org/42/291142/1/gate/gate-instack-undercloud-pep8/bedc19b/ : SUCCESS in 2m 30s\n- gate-instack-undercloud-python27 http://logs.openstack.org/42/291142/1/gate/gate-instack-undercloud-python27/96726b8/ : SUCCESS in 1m 36s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/42/291142/1/gate/gate-instack-undercloud-puppet-lint/42fc1b7/ : SUCCESS in 1m 23s\n- gate-instack-undercloud-puppet-syntax-3-dsvm-centos7 http://logs.openstack.org/42/291142/1/gate/gate-instack-undercloud-puppet-syntax-3-dsvm-centos7/de711fe/ : SUCCESS in 50s\n- gate-instack-undercloud-puppet-syntax-4-dsvm-centos7 http://logs.openstack.org/42/291142/1/gate/gate-instack-undercloud-puppet-syntax-4-dsvm-centos7/2cac5ac/ : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62efd77fd5c5838ab2b2b741e5e21bd6a7bc6a79","revisions":{"62efd77fd5c5838ab2b2b741e5e21bd6a7bc6a79":{"kind":"REWORK","_number":1,"created":"2016-03-10 12:25:53.000000000","uploader":{"_account_id":9979,"name":"Mike Burns","email":"mburns@redhat.com","username":"mburns"},"ref":"refs/changes/42/291142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/42/291142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/42/291142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/42/291142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/42/291142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/42/291142/1"}}},"commit":{"parents":[{"commit":"c2ed2f2a98d00479b52ece4de5b0bcdf065ca71e","subject":"Store events in Undercloud Ceilometer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/c2ed2f2a98d00479b52ece4de5b0bcdf065ca71e"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2015-10-06 12:44:43.000000000","tz":-240},"committer":{"name":"Mike Burns","email":"mburns@redhat.com","date":"2016-03-10 12:25:53.000000000","tz":0},"subject":"Use service tenant for ceilometer","message":"Use service tenant for ceilometer\n\nConfigure ceilometer to use the service tenant instead of the admin\ntenant.  Using the admin tenant is not required and a security risk.\n\nThis brings the ceilometer configuration in line with the\nrecommendations from the official installation guide:\nhttp://docs.openstack.org/kilo/install-guide/install/yum/content/ceilometer-controller-install.html\n\nChange-Id: I40dd4543580332a46cde96f4bca639eb777ac7a8\nPartial-Bug: #1358237\n(cherry picked from commit 99b132b0ff258b1b7fc8eca0cbeec69613b59175)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/62efd77fd5c5838ab2b2b741e5e21bd6a7bc6a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/62efd77fd5c5838ab2b2b741e5e21bd6a7bc6a79"}]},"parents_data":[{"branch_name":"refs/heads/stable/liberty","commit_id":"c2ed2f2a98d00479b52ece4de5b0bcdf065ca71e","is_merged_in_target_branch":true,"change_id":"I9d9ae8eb2e1ec069e2ecd0bcf553f57647750ef1","change_number":289789,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/liberty"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
