)]}'
{"id":"openstack%2Finstack-undercloud~375054","triplet_id":"openstack%2Finstack-undercloud~master~I435405d787134b1f0ea3e6bfef4b66faa9fde551","project":"openstack/instack-undercloud","branch":"master","hashtags":[],"change_id":"I435405d787134b1f0ea3e6bfef4b66faa9fde551","subject":"Update default VM memory","status":"MERGED","created":"2016-09-22 18:45:13.000000000","updated":"2016-09-23 03:02:47.000000000","submitted":"2016-09-23 03:02:47.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bdeebd0018d6a2388ce4f9e6e3391ea827bd4328","_number":375054,"virtual_id_number":375054,"owner":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2016-09-23 03:02:47.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2016-09-22 18:49:47.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-09-22 22:12:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2016-09-23 01:28:05.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-09-22 18:49:47.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2016-09-22 22:12:26.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2016-09-23 01:28:05.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2016-09-23 03:02:47.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"40dd27e666bd2369018bdce84b7c69e49fabe7c1","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2016-09-22 18:45:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ff7bd244942b159bc78487040d94e9857657af4","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2016-09-22 18:49:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"649639dc4f5bbe61731457e116494fffdc9fb6f5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-22 20:08:31.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/54/375054/1/check/gate-instack-undercloud-docs-ubuntu-xenial/4b71432//doc/build/html/ : SUCCESS in 2m 22s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/54/375054/1/check/gate-instack-undercloud-pep8-ubuntu-xenial/0ed21d8/ : SUCCESS in 3m 48s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/54/375054/1/check/gate-instack-undercloud-python27-ubuntu-xenial/dfa6604/ : SUCCESS in 3m 26s\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/54/375054/1/check/gate-tripleo-ci-centos-7-nonha-multinode/e2e6600/ : SUCCESS in 1h 15m 06s\n- gate-tripleo-ci-centos-7-undercloud http://logs.openstack.org/54/375054/1/check/gate-tripleo-ci-centos-7-undercloud/2f3e10d/ : SUCCESS in 40m 53s\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/54/375054/1/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/c4e6f57/ : FAILURE in 1h 16m 08s (non-voting)\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/54/375054/1/check/gate-instack-undercloud-puppet-lint/f431c1c/ : FAILURE in 3m 17s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/54/375054/1/check/gate-instack-undercloud-puppet-syntax-3-centos-7/de601c1/ : SUCCESS in 5m 42s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/54/375054/1/check/gate-instack-undercloud-puppet-syntax-4-centos-7/8b5eb36/ : SUCCESS in 7m 43s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/54/375054/1/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/cbc8eaf/ : SUCCESS in 27m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3820977c6c592db36eb6b22b40fc8c7cfc71669b","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2016-09-22 20:14:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"305d6f34df920ec8e8e6870d29f045b1adbbb786","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-22 20:57:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha http://logs.openstack.org/54/375054/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha/b9eceb6/ : SUCCESS in 2h 10m 14s\n- gate-tripleo-ci-centos-7-ovb-nonha http://logs.openstack.org/54/375054/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-nonha/924b019/ : SUCCESS in 2h 00m 28s","accounts_in_message":[],"_revision_number":1},{"id":"38542bac9ed71b7cd8b5a8256c848d99ee5cf231","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-22 21:37:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/54/375054/1/check/gate-instack-undercloud-docs-ubuntu-xenial/280a47f//doc/build/html/ : SUCCESS in 2m 42s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/54/375054/1/check/gate-instack-undercloud-pep8-ubuntu-xenial/59a6b07/ : SUCCESS in 2m 29s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/54/375054/1/check/gate-instack-undercloud-python27-ubuntu-xenial/e5627d1/ : SUCCESS in 2m 42s\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/54/375054/1/check/gate-tripleo-ci-centos-7-nonha-multinode/8268d28/ : SUCCESS in 1h 17m 26s\n- gate-tripleo-ci-centos-7-undercloud http://logs.openstack.org/54/375054/1/check/gate-tripleo-ci-centos-7-undercloud/3b47fb6/ : SUCCESS in 19m 11s\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/54/375054/1/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/8238fdf/ : FAILURE in 1h 07m 47s (non-voting)\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/54/375054/1/check/gate-instack-undercloud-puppet-lint/d8c5425/ : SUCCESS in 2m 30s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/54/375054/1/check/gate-instack-undercloud-puppet-syntax-3-centos-7/744b9f8/ : SUCCESS in 3m 30s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/54/375054/1/check/gate-instack-undercloud-puppet-syntax-4-centos-7/29bf564/ : SUCCESS in 4m 40s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/54/375054/1/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/66f3ece/ : SUCCESS in 34m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2615ef96d566029954ed937c8b9daba39a1720a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2016-09-22 22:12:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1c4321440bfa6dac993a379aa2e478d54fc107b8","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2016-09-23 01:28:05.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"73c913209c8391524065733a0ba4201f753a231e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-23 01:28:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"260a5bf2827ec8a4c1dce24834ad54d501c26eb4","date":"2016-09-23 03:02:47.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"bdeebd0018d6a2388ce4f9e6e3391ea827bd4328","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-09-23 03:02:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/54/375054/1/gate/gate-instack-undercloud-docs-ubuntu-xenial/ff0968c//doc/build/html/ : SUCCESS in 3m 22s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/54/375054/1/gate/gate-instack-undercloud-pep8-ubuntu-xenial/c8dbaa3/ : SUCCESS in 3m 19s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/54/375054/1/gate/gate-instack-undercloud-python27-ubuntu-xenial/11b9627/ : SUCCESS in 3m 43s\n- gate-tripleo-ci-centos-7-nonha-multinode http://logs.openstack.org/54/375054/1/gate/gate-tripleo-ci-centos-7-nonha-multinode/bd722c8/ : SUCCESS in 1h 03m 01s\n- gate-tripleo-ci-centos-7-undercloud http://logs.openstack.org/54/375054/1/gate/gate-tripleo-ci-centos-7-undercloud/2d9d085/ : SUCCESS in 21m 39s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/54/375054/1/gate/gate-instack-undercloud-puppet-lint/175c3f4/ : SUCCESS in 2m 08s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/54/375054/1/gate/gate-instack-undercloud-puppet-syntax-3-centos-7/0eab09d/ : SUCCESS in 5m 09s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/54/375054/1/gate/gate-instack-undercloud-puppet-syntax-4-centos-7/01e0a7e/ : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cea03d4a5f020422998660f40e05f2364530cf66","revisions":{"cea03d4a5f020422998660f40e05f2364530cf66":{"kind":"REWORK","_number":1,"created":"2016-09-22 18:45:13.000000000","uploader":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"ref":"refs/changes/54/375054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/54/375054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/54/375054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/54/375054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/54/375054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/54/375054/1"}}},"commit":{"parents":[{"commit":"bdde348c331bf3588f0984497f4ab67e9226f574","subject":"Merge \"Use correct password for keystone bootstrap\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/bdde348c331bf3588f0984497f4ab67e9226f574"}]}],"author":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2016-09-22 18:44:09.000000000","tz":-240},"committer":{"name":"James Slagle","email":"jslagle@redhat.com","date":"2016-09-22 18:44:09.000000000","tz":-240},"subject":"Update default VM memory","message":"Update default VM memory\n\nUpdate instack-virt-setup to match the known minimum requirements for VM\nmemory (matches what tripleo-ci currently uses).\n\nChange-Id: I435405d787134b1f0ea3e6bfef4b66faa9fde551\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/cea03d4a5f020422998660f40e05f2364530cf66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/cea03d4a5f020422998660f40e05f2364530cf66"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"bdde348c331bf3588f0984497f4ab67e9226f574","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
