)]}'
{"id":"openstack%2Finstack-undercloud~453181","triplet_id":"openstack%2Finstack-undercloud~stable%2Focata~I74bc5a60a25a8b21567a3bd7edd5b4208379cb8a","project":"openstack/instack-undercloud","branch":"stable/ocata","topic":"test-collector","hashtags":[],"change_id":"I74bc5a60a25a8b21567a3bd7edd5b4208379cb8a","subject":"TEST: Set a dependency on collector","status":"ABANDONED","created":"2017-04-04 13:56:37.000000000","updated":"2017-04-06 14:09:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5d5b76d29e1730328622f83982418f4e3453b0a6","_number":453181,"virtual_id_number":453181,"owner":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":1,"date":"2017-04-05 20:38:01.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3,"name":"Jenkins","username":"jenkins"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-04-05 20:38:01.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"37d612c7f56ad2487f0091f9505f5270213ffb89","author":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"date":"2017-04-04 13:56:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c866933251c6b00efabd5f0414766443e598cf9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-04 15:12:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/81/453181/1/check/gate-instack-undercloud-docs-ubuntu-xenial/e1d3a19//doc/build/html/ : SUCCESS in 45s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/81/453181/1/check/gate-instack-undercloud-pep8-ubuntu-xenial/1def663/ : SUCCESS in 1m 25s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/81/453181/1/check/gate-instack-undercloud-python27-ubuntu-xenial/1a8472f/ : SUCCESS in 1m 43s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/81/453181/1/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/34de72e/ : SUCCESS in 1h 05m 45s\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/81/453181/1/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/0765413/ : SUCCESS in 1h 13m 14s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/81/453181/1/check/gate-tripleo-ci-centos-7-undercloud-oooq/e498d98/ : SUCCESS in 43m 10s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/81/453181/1/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/4d8ef5c/ : FAILURE in 58m 46s (non-voting)\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/81/453181/1/check/gate-instack-undercloud-puppet-lint/9d90560/ : SUCCESS in 1m 55s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/81/453181/1/check/gate-instack-undercloud-puppet-syntax-3-centos-7/64e5c71/ : SUCCESS in 1m 48s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/81/453181/1/check/gate-instack-undercloud-puppet-syntax-4-centos-7/2efc57b/ : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1146c79ecac03982af317e076620b3b1656ab7f3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-04 17:14:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha http://logs.openstack.org/81/453181/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha/33f2064/ : SUCCESS in 2h 26m 07s\n- gate-tripleo-ci-centos-7-ovb-nonha http://logs.openstack.org/81/453181/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-nonha/5c51588/ : SUCCESS in 2h 04m 35s\n- gate-tripleo-ci-centos-7-ovb-updates http://logs.openstack.org/81/453181/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-updates/b1325db/ : SUCCESS in 2h 18m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5f84f52bd1c11c8f1a86ad99d7d3ceed66c1156b","author":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"date":"2017-04-04 21:24:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f1e25b49cbfd4fabe4ac23162a80dd11c24b14f6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-04 22:53:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/81/453181/2/check/gate-instack-undercloud-docs-ubuntu-xenial/5b56036//doc/build/html/ : SUCCESS in 1m 36s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/81/453181/2/check/gate-instack-undercloud-pep8-ubuntu-xenial/75989d8/ : SUCCESS in 1m 41s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/81/453181/2/check/gate-instack-undercloud-python27-ubuntu-xenial/2b6b6c6/ : SUCCESS in 1m 44s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/81/453181/2/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/d2da2c9/ : SUCCESS in 1h 29m 27s\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/81/453181/2/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/9e46b51/ : SUCCESS in 1h 12m 48s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/81/453181/2/check/gate-tripleo-ci-centos-7-undercloud-oooq/057e98a/ : SUCCESS in 54m 59s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/81/453181/2/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/3b8e963/ : FAILURE in 55m 52s (non-voting)\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/81/453181/2/check/gate-instack-undercloud-puppet-lint/0386dbf/ : SUCCESS in 2m 10s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/81/453181/2/check/gate-instack-undercloud-puppet-syntax-3-centos-7/8453b11/ : SUCCESS in 2m 16s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/81/453181/2/check/gate-instack-undercloud-puppet-syntax-4-centos-7/e68e427/ : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":2},{"id":"5e770c967dd2408dd1c40d6a58e371dc141b664e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-05 01:40:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha http://logs.openstack.org/81/453181/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha/d2a2aca/ : SUCCESS in 2h 21m 56s\n- gate-tripleo-ci-centos-7-ovb-nonha http://logs.openstack.org/81/453181/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-nonha/78e7da2/ : SUCCESS in 2h 07m 31s\n- gate-tripleo-ci-centos-7-ovb-updates http://logs.openstack.org/81/453181/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-updates/1e4de33/ : SUCCESS in 2h 16m 04s","accounts_in_message":[],"_revision_number":2},{"id":"224c76fd191c15365c0efa289646d3001c81229d","author":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"date":"2017-04-05 19:11:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"15267a2990e316977962d58296e40ca5040503cf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-05 20:38:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/81/453181/3/check/gate-instack-undercloud-docs-ubuntu-xenial/9550d34//doc/build/html/ : SUCCESS in 1m 32s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/81/453181/3/check/gate-instack-undercloud-pep8-ubuntu-xenial/bbe9c02/ : SUCCESS in 2m 56s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/81/453181/3/check/gate-instack-undercloud-python27-ubuntu-xenial/f523bde/ : SUCCESS in 1m 31s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/81/453181/3/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/73a1a7f/ : SUCCESS in 1h 25m 56s\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/81/453181/3/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/f81bcaf/ : SUCCESS in 1h 12m 41s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/81/453181/3/check/gate-tripleo-ci-centos-7-undercloud-oooq/5e5b8ac/ : SUCCESS in 41m 59s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/81/453181/3/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/13051f8/ : FAILURE in 41m 31s (non-voting)\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/81/453181/3/check/gate-instack-undercloud-puppet-lint/205bf81/ : SUCCESS in 3m 59s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/81/453181/3/check/gate-instack-undercloud-puppet-syntax-3-centos-7/dcc8d22/ : SUCCESS in 1m 56s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/81/453181/3/check/gate-instack-undercloud-puppet-syntax-4-centos-7/298eda7/ : SUCCESS in 2m 18s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/81/453181/3/check/instack-undercloud-coverage-ubuntu-xenial-nv/b3a3f73/ : SUCCESS in 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ba297481123d538c3889c264a714f5a36e9074e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-04-05 22:09:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha http://logs.openstack.org/81/453181/3/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha/a487f23/ : SUCCESS in 2h 19m 52s\n- gate-tripleo-ci-centos-7-ovb-nonha http://logs.openstack.org/81/453181/3/check-tripleo/gate-tripleo-ci-centos-7-ovb-nonha/21fc968/ : SUCCESS in 1h 58m 07s\n- gate-tripleo-ci-centos-7-ovb-updates http://logs.openstack.org/81/453181/3/check-tripleo/gate-tripleo-ci-centos-7-ovb-updates/ee21cc1/ : SUCCESS in 2h 18m 07s","accounts_in_message":[],"_revision_number":3},{"id":"52c9900561b73eee896f96906bed192f6fcc45f6","author":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"date":"2017-04-06 14:09:12.000000000","message":"Abandoned\n\ncherry-picked from master https://review.openstack.org/#/c/454119/","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8eedacea3e516681d8ea51694c48e84d81d95cda","revisions":{"e6e5535a63d029c3175c8e1be1d29778cf32edaa":{"kind":"REWORK","_number":1,"created":"2017-04-04 13:56:37.000000000","uploader":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"ref":"refs/changes/81/453181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/81/453181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/81/453181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/81/453181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/81/453181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/81/453181/1"}}},"commit":{"parents":[{"commit":"7aff3cdecb25d47b6e01be4894470ff8ca7605d3","subject":"Remove obsolete flag on gnocchi-upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/7aff3cdecb25d47b6e01be4894470ff8ca7605d3"}]}],"author":{"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","date":"2017-04-04 13:52:04.000000000","tz":-240},"committer":{"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","date":"2017-04-04 13:52:04.000000000","tz":-240},"subject":"TEST: Set a dependency on collector","message":"TEST: Set a dependency on collector\n\nChange-Id: I74bc5a60a25a8b21567a3bd7edd5b4208379cb8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/e6e5535a63d029c3175c8e1be1d29778cf32edaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/e6e5535a63d029c3175c8e1be1d29778cf32edaa"}]},"parents_data":[{"branch_name":"refs/heads/stable/ocata","commit_id":"7aff3cdecb25d47b6e01be4894470ff8ca7605d3","is_merged_in_target_branch":true,"change_id":"I4f877473f2b4b8d58a57769e1ee563282dbe28df","change_number":452912,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/ocata"},"177d574f1faf09c389e7183b5bb1fa61030c7a10":{"kind":"REWORK","_number":2,"created":"2017-04-04 21:24:10.000000000","uploader":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"ref":"refs/changes/81/453181/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/81/453181/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/81/453181/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/81/453181/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/81/453181/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/81/453181/2"}}},"commit":{"parents":[{"commit":"7aff3cdecb25d47b6e01be4894470ff8ca7605d3","subject":"Remove obsolete flag on gnocchi-upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/7aff3cdecb25d47b6e01be4894470ff8ca7605d3"}]}],"author":{"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","date":"2017-04-04 13:52:04.000000000","tz":-240},"committer":{"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","date":"2017-04-04 21:19:06.000000000","tz":-240},"subject":"TEST: Set a dependency on collector","message":"TEST: Set a dependency on collector\n\nChange-Id: I74bc5a60a25a8b21567a3bd7edd5b4208379cb8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/177d574f1faf09c389e7183b5bb1fa61030c7a10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/177d574f1faf09c389e7183b5bb1fa61030c7a10"}]},"parents_data":[{"branch_name":"refs/heads/stable/ocata","commit_id":"7aff3cdecb25d47b6e01be4894470ff8ca7605d3","is_merged_in_target_branch":true,"change_id":"I4f877473f2b4b8d58a57769e1ee563282dbe28df","change_number":452912,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/ocata"},"8eedacea3e516681d8ea51694c48e84d81d95cda":{"kind":"REWORK","_number":3,"created":"2017-04-05 19:11:03.000000000","uploader":{"_account_id":6924,"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","username":"pkilambi"},"ref":"refs/changes/81/453181/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/81/453181/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/81/453181/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/81/453181/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/81/453181/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/81/453181/3"}}},"commit":{"parents":[{"commit":"7aff3cdecb25d47b6e01be4894470ff8ca7605d3","subject":"Remove obsolete flag on gnocchi-upgrade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/7aff3cdecb25d47b6e01be4894470ff8ca7605d3"}]}],"author":{"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","date":"2017-04-04 13:52:04.000000000","tz":-240},"committer":{"name":"Pradeep Kilambi","email":"pkilambi@redhat.com","date":"2017-04-05 19:06:32.000000000","tz":-240},"subject":"TEST: Set a dependency on collector","message":"TEST: Set a dependency on collector\n\nChange-Id: I74bc5a60a25a8b21567a3bd7edd5b4208379cb8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/8eedacea3e516681d8ea51694c48e84d81d95cda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/8eedacea3e516681d8ea51694c48e84d81d95cda"}]},"parents_data":[{"branch_name":"refs/heads/stable/ocata","commit_id":"7aff3cdecb25d47b6e01be4894470ff8ca7605d3","is_merged_in_target_branch":true,"change_id":"I4f877473f2b4b8d58a57769e1ee563282dbe28df","change_number":452912,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
