)]}'
{"id":"openstack%2Finstack-undercloud~470007","triplet_id":"openstack%2Finstack-undercloud~master~I61e5fc005aa27ba76f91a2b84fec751af93061b5","project":"openstack/instack-undercloud","branch":"master","topic":"bug/1698010","hashtags":[],"change_id":"I61e5fc005aa27ba76f91a2b84fec751af93061b5","subject":"Allow setting the DNS domain name in undercloud.conf","status":"MERGED","created":"2017-06-01 20:49:37.000000000","updated":"2017-06-15 16:09:28.000000000","submitted":"2017-06-15 16:09:28.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"35d843e16a843ba0fb5197ef40fb8e73a4e2c59b","_number":470007,"virtual_id_number":470007,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2017-06-15 16:09:28.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"recommended":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-06-15 11:07:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":2,"date":"2017-06-15 12:33:30.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2017-06-15 06:18:39.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},{"value":1,"date":"2017-06-15 12:33:30.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-06-15 06:18:39.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2017-06-15 11:07:28.000000000","updated_by":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"reviewer":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"state":"REVIEWER"},{"updated":"2017-06-15 12:33:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2017-06-15 16:09:28.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"5ff85d44e5136dd34f03dc055fb617a65991868e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2017-06-01 20:49:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20fbc375c8853dfcc680dccf99c370342ce63fc1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-01 22:22:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/07/470007/1/check/gate-instack-undercloud-docs-ubuntu-xenial/76caa81//doc/build/html/ : SUCCESS in 1m 24s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/07/470007/1/check/gate-instack-undercloud-pep8-ubuntu-xenial/f1d0a0e/ : SUCCESS in 2m 22s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/07/470007/1/check/gate-instack-undercloud-python27-ubuntu-xenial/92dd28d/ : SUCCESS in 1m 23s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/07/470007/1/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/2a42d40/ : SUCCESS in 1h 25m 32s\n- gate-tripleo-ci-centos-7-containers-multinode-nv http://logs.openstack.org/07/470007/1/check/gate-tripleo-ci-centos-7-containers-multinode-nv/ff02c0c/ : SUCCESS in 1h 31m 40s (non-voting)\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/07/470007/1/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/fbd2563/ : SUCCESS in 1h 21m 38s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/07/470007/1/check/gate-tripleo-ci-centos-7-undercloud-oooq/2ba0c5e/ : SUCCESS in 1h 24m 29s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/07/470007/1/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/0bf19e0/ : SUCCESS in 41m 27s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-containers-nv http://logs.openstack.org/07/470007/1/check/gate-tripleo-ci-centos-7-undercloud-containers-nv/8805545/ : FAILURE in 30m 14s (non-voting)\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/07/470007/1/check/gate-instack-undercloud-puppet-lint/4a410ad/ : SUCCESS in 2m 55s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/07/470007/1/check/gate-instack-undercloud-puppet-syntax-3-centos-7/26eb063/ : SUCCESS in 2m 23s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/07/470007/1/check/gate-instack-undercloud-puppet-syntax-4-centos-7/5d6ad71/ : SUCCESS in 2m 46s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/07/470007/1/check/instack-undercloud-coverage-ubuntu-xenial-nv/2218b5c/ : SUCCESS in 1m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"959b1c8616fd9849bec8e529fe96251d3a56725a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-01 23:35:26.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/07/470007/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/837b34d/ : FAILURE in 2h 45m 19s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq-nv http://logs.openstack.org/07/470007/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq-nv/7c5df81/ : FAILURE in 2h 25m 59s (non-voting)\n- gate-tripleo-ci-centos-7-ovb-updates http://logs.openstack.org/07/470007/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-updates/012dde7/ : FAILURE in 2h 41m 18s","accounts_in_message":[],"_revision_number":1},{"id":"9ba68bbeef8dd8c4417ecf089b57f53f27793687","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2017-06-02 20:28:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6a717c5e35b1ba23540d6101f9171bebc76b8840","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-06-02 20:41:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27250d7babd16923b9bdaeaf10a05e77e59412a5","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2017-06-05 10:30:57.000000000","message":"Patch Set 1: Code-Review-1\n\nLGTM, but please create a release note (especially mentioning interaction between CloudDomain and this new option)","accounts_in_message":[],"_revision_number":1},{"id":"bff7e9e27833c846b40b5934f5d0a1899fa9ee2f","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2017-06-14 21:54:53.000000000","message":"Patch Set 1:\n\n\u003e LGTM, but please create a release note (especially mentioning\n \u003e interaction between CloudDomain and this new option)\n\nWill do.  It looks like we also need to set Nova\u0027s dhcp_domain option for this to work as expected (despite that opt being deprecated...).  I\u0027ve opened a bug for that: https://bugs.launchpad.net/nova/+bug/1698010","accounts_in_message":[],"_revision_number":1},{"id":"44da9feb6282d39072be0b303296abab04234683","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2017-06-14 21:58:34.000000000","message":"Patch Set 1: Workflow-1\n\nOh, I also forgot to update the sample config. :-/","accounts_in_message":[],"_revision_number":1},{"id":"e40638365da352911aad4786c5bffcc835cc1161","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2017-06-14 22:03:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"838bc1964c0da706c4bbc634612792cbb6aa908e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-15 00:45:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/07/470007/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/727aae5/ : SUCCESS in 2h 41m 09s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq-nv http://logs.openstack.org/07/470007/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq-nv/18e0a45/ : FAILURE in 2h 40m 46s (non-voting)\n- gate-tripleo-ci-centos-7-ovb-updates http://logs.openstack.org/07/470007/2/check-tripleo/gate-tripleo-ci-centos-7-ovb-updates/c59f97b/ : SUCCESS in 2h 23m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c43378f8de55128c5bc7207d60d51f6ac5c44b9e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-15 00:52:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/07/470007/2/check/gate-instack-undercloud-docs-ubuntu-xenial/e6154eb//doc/build/html/ : SUCCESS in 2m 11s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/07/470007/2/check/gate-instack-undercloud-pep8-ubuntu-xenial/31e0b3a/ : SUCCESS in 1m 44s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/07/470007/2/check/gate-instack-undercloud-python27-ubuntu-xenial/f63802b/ : SUCCESS in 1m 25s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/07/470007/2/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/1092668/ : SUCCESS in 1h 12m 51s\n- gate-tripleo-ci-centos-7-containers-multinode-nv http://logs.openstack.org/07/470007/2/check/gate-tripleo-ci-centos-7-containers-multinode-nv/f33bedc/ : FAILURE in 2h 47m 32s (non-voting)\n- gate-tripleo-ci-centos-7-nonha-multinode-updates-nv http://logs.openstack.org/07/470007/2/check/gate-tripleo-ci-centos-7-nonha-multinode-updates-nv/7134f88/ : SUCCESS in 1h 26m 10s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/07/470007/2/check/gate-tripleo-ci-centos-7-undercloud-oooq/c3a86fc/ : SUCCESS in 1h 07m 53s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/07/470007/2/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/8fed211/ : SUCCESS in 44m 27s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-containers-nv http://logs.openstack.org/07/470007/2/check/gate-tripleo-ci-centos-7-undercloud-containers-nv/f1ce81e/ : SUCCESS in 31m 28s (non-voting)\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/07/470007/2/check/gate-instack-undercloud-releasenotes/87d15d1//releasenotes/build/html/ : SUCCESS in 2m 33s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/07/470007/2/check/gate-instack-undercloud-puppet-lint/38b94d0/ : SUCCESS in 1m 25s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/07/470007/2/check/gate-instack-undercloud-puppet-syntax-3-centos-7/6749c95/ : SUCCESS in 3m 04s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/07/470007/2/check/gate-instack-undercloud-puppet-syntax-4-centos-7/b4fcc12/ : SUCCESS in 2m 18s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/07/470007/2/check/instack-undercloud-coverage-ubuntu-xenial-nv/ead333b/ : SUCCESS in 1m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7173bdfdb784c55fc8489f82bababaea5ffadca9","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2017-06-15 06:18:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a4f06e5ec19300881a4e014db58d156e3e2cb75c","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2017-06-15 11:07:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b326455324251db9de6fee27b6fdb8621e423aa9","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2017-06-15 12:33:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c2ad171860b12b570f96b7d0c2e3245332838925","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-15 12:33:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"ce17e6f746c109701e9fad38a71cff3f3eb62c16","date":"2017-06-15 16:09:28.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":2},{"id":"79869f4e2d5064884662c1a18a45fdb4af2b5e8e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-06-15 16:09:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/07/470007/2/gate/gate-instack-undercloud-docs-ubuntu-xenial/2e3b271//doc/build/html/ : SUCCESS in 1m 24s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/07/470007/2/gate/gate-instack-undercloud-pep8-ubuntu-xenial/c63a279/ : SUCCESS in 2m 25s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/07/470007/2/gate/gate-instack-undercloud-python27-ubuntu-xenial/79b7f79/ : SUCCESS in 3m 54s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/07/470007/2/gate/gate-tripleo-ci-centos-7-nonha-multinode-oooq/50608e0/ : SUCCESS in 1h 32m 37s\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/07/470007/2/gate/gate-tripleo-ci-centos-7-undercloud-oooq/b562a09/ : SUCCESS in 42m 37s\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/07/470007/2/gate/gate-instack-undercloud-releasenotes/30c8ad1//releasenotes/build/html/ : SUCCESS in 58s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/07/470007/2/gate/gate-instack-undercloud-puppet-lint/efa7ed1/ : SUCCESS in 2m 56s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/07/470007/2/gate/gate-instack-undercloud-puppet-syntax-3-centos-7/878e8a6/ : SUCCESS in 2m 28s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/07/470007/2/gate/gate-instack-undercloud-puppet-syntax-4-centos-7/b75761a/ : SUCCESS in 3m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"675a59e8cec2527b9e1e78654a616b2672328021","revisions":{"0d2aedfe8248dd77e99ba6de821a45c61289775d":{"kind":"REWORK","_number":1,"created":"2017-06-01 20:49:37.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/07/470007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/07/470007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/07/470007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/07/470007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/07/470007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/07/470007/1"}}},"commit":{"parents":[{"commit":"5903151ff5f42ee9e3a3e0a58d17df4a86fb2da8","subject":"Log exception tracebacks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/5903151ff5f42ee9e3a3e0a58d17df4a86fb2da8"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2017-06-01 20:38:09.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2017-06-01 20:38:09.000000000","tz":0},"subject":"Allow setting the DNS domain name in undercloud.conf","message":"Allow setting the DNS domain name in undercloud.conf\n\nWithout being able to do this, users who want to set CloudDomain\nduring their deployment will have to manually override neutron\u0027s\ndns_domain config, which either requires custom hieradata or the\nreapplication of the customization after each undercloud update.\nThis is not a good user experience, so we should allow them to set\nit once in the config file and never have to touch it again.\n\nChange-Id: I61e5fc005aa27ba76f91a2b84fec751af93061b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/0d2aedfe8248dd77e99ba6de821a45c61289775d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/0d2aedfe8248dd77e99ba6de821a45c61289775d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5903151ff5f42ee9e3a3e0a58d17df4a86fb2da8","is_merged_in_target_branch":true,"change_id":"Ifbc273b9c5a7de42aed8093d563d514348ddeaa3","change_number":468426,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"675a59e8cec2527b9e1e78654a616b2672328021":{"kind":"REWORK","_number":2,"created":"2017-06-14 22:03:38.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/07/470007/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/07/470007/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/07/470007/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/07/470007/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/07/470007/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/07/470007/2"}}},"commit":{"parents":[{"commit":"5903151ff5f42ee9e3a3e0a58d17df4a86fb2da8","subject":"Log exception tracebacks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/5903151ff5f42ee9e3a3e0a58d17df4a86fb2da8"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2017-06-01 20:38:09.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2017-06-14 22:02:19.000000000","tz":0},"subject":"Allow setting the DNS domain name in undercloud.conf","message":"Allow setting the DNS domain name in undercloud.conf\n\nWithout being able to do this, users who want to set CloudDomain\nduring their deployment will have to manually override neutron\u0027s\ndns_domain config, which either requires custom hieradata or the\nreapplication of the customization after each undercloud update.\nThis is not a good user experience, so we should allow them to set\nit once in the config file and never have to touch it again.\n\nIt turns out that the Nova dhcp_domain option must also be set,\neven though it is a deprecated option.  A bug with nova has been\nopened to address this in the future.\n\nChange-Id: I61e5fc005aa27ba76f91a2b84fec751af93061b5\nRelated-Bug: 1698010\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/675a59e8cec2527b9e1e78654a616b2672328021"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/675a59e8cec2527b9e1e78654a616b2672328021"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5903151ff5f42ee9e3a3e0a58d17df4a86fb2da8","is_merged_in_target_branch":true,"change_id":"Ifbc273b9c5a7de42aed8093d563d514348ddeaa3","change_number":468426,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
