)]}'
{"id":"openstack%2Finstack-undercloud~503627","triplet_id":"openstack%2Finstack-undercloud~master~I78c77c005a22f97613e13a6b2b3f7ab73ae01859","project":"openstack/instack-undercloud","branch":"master","topic":"openstack/requirements","hashtags":[],"change_id":"I78c77c005a22f97613e13a6b2b3f7ab73ae01859","subject":"Updated from global requirements","status":"MERGED","created":"2017-09-13 12:51:11.000000000","updated":"2017-09-22 19:53:53.000000000","submitted":"2017-09-22 19:53:53.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"503627-1506110033096-8e14b532","meta_rev_id":"ab011af4a9b167e7e5d8a4ccf7a8198f13471afb","_number":503627,"virtual_id_number":503627,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":25747,"name":"Debo Zhang","email":"zhangdebo@inspur.com","username":"zhangdebo"},{"value":2,"date":"2017-09-22 19:53:52.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"date":"2017-09-22 15:55:11.000000000","_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2017-09-22 13:30:43.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2017-09-22 12:51:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25747,"name":"Debo Zhang","email":"zhangdebo@inspur.com","username":"zhangdebo"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-09-22 12:51:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":2,"date":"2017-09-22 12:51:36.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2017-09-22 13:30:43.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":25747,"name":"Debo Zhang","email":"zhangdebo@inspur.com","username":"zhangdebo"},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"value":0,"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":25747,"name":"Debo Zhang","email":"zhangdebo@inspur.com","username":"zhangdebo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-09-18 09:50:08.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2017-09-20 14:40:14.000000000","updated_by":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"reviewer":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"state":"REVIEWER"},{"updated":"2017-09-21 06:26:18.000000000","updated_by":{"_account_id":25747,"name":"Debo Zhang","email":"zhangdebo@inspur.com","username":"zhangdebo"},"reviewer":{"_account_id":25747,"name":"Debo Zhang","email":"zhangdebo@inspur.com","username":"zhangdebo"},"state":"REVIEWER"},{"updated":"2017-09-22 13:30:43.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2017-09-22 15:55:11.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2017-09-22 19:53:52.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"e7db39b54e88a3cc9fc73bc5ef82cc4be8903534","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-13 12:51:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9b1b70f19fb6b5cef8b1040ebd45abf454b77c7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-13 12:59:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/27/503627/1/check/gate-instack-undercloud-docs-ubuntu-xenial/7129273//doc/build/html/ : SUCCESS in 2m 41s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/27/503627/1/check/gate-instack-undercloud-pep8-ubuntu-xenial/5bcc2da/ : SUCCESS in 3m 30s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/27/503627/1/check/gate-instack-undercloud-python27-ubuntu-xenial/ec56c90/ : SUCCESS in 3m 28s\n- gate-instack-undercloud-python35 http://logs.openstack.org/27/503627/1/check/gate-instack-undercloud-python35/c22be29/ : SUCCESS in 2m 26s\n- gate-instack-undercloud-requirements http://logs.openstack.org/27/503627/1/check/gate-instack-undercloud-requirements/057f5d7/ : SUCCESS in 2m 06s\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/27/503627/1/check/gate-instack-undercloud-releasenotes/e74f0e7//releasenotes/build/html/ : SUCCESS in 1m 51s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/27/503627/1/check/gate-instack-undercloud-puppet-syntax-3-centos-7/a7a7d3c/ : SUCCESS in 3m 56s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/27/503627/1/check/gate-instack-undercloud-puppet-syntax-4-centos-7/a53b4a3/ : SUCCESS in 3m 33s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/27/503627/1/check/gate-instack-undercloud-puppet-lint/1297be0/ : SUCCESS in 3m 52s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/27/503627/1/check/instack-undercloud-coverage-ubuntu-xenial-nv/0b232f0/ : SUCCESS in 2m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"32f377eb30207aee558d645b277c0f2ad5ffe6ed","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-13 15:21:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/27/503627/1/check-tripleo/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/213558b/ : SUCCESS in 2h 30m 05s","accounts_in_message":[],"_revision_number":1},{"id":"fac48fc8f10dbd95e71c55ad483815d3fa92ab46","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2017-09-14 05:03:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b9bcce18c0802230f3d603fe7e69e234c7e0862b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-16 23:12:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"40b6ca966e34652bfdba80616fe9ed4885b0af36","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-17 02:56:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/27/503627/2/check/gate-instack-undercloud-docs-ubuntu-xenial/adf0e7b//doc/build/html/ : SUCCESS in 2m 21s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/27/503627/2/check/gate-instack-undercloud-pep8-ubuntu-xenial/4675574/ : SUCCESS in 2m 36s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/27/503627/2/check/gate-instack-undercloud-python27-ubuntu-xenial/ba47deb/ : SUCCESS in 2m 10s\n- gate-instack-undercloud-python35 http://logs.openstack.org/27/503627/2/check/gate-instack-undercloud-python35/5f98e88/ : SUCCESS in 2m 03s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/27/503627/2/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/aa4e12b/ : FAILURE in 1h 47m 55s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/27/503627/2/check/gate-tripleo-ci-centos-7-containers-multinode/6258331/ : SUCCESS in 1h 44m 55s\n- gate-tripleo-ci-centos-7-undercloud-containers-nv http://logs.openstack.org/27/503627/2/check/gate-tripleo-ci-centos-7-undercloud-containers-nv/d33703e/ : FAILURE in 37m 44s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/27/503627/2/check/gate-tripleo-ci-centos-7-undercloud-oooq/84fbf0d/ : SUCCESS in 1h 08m 16s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/27/503627/2/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/3f59fc6/ : SUCCESS in 54m 36s (non-voting)\n- gate-instack-undercloud-requirements http://logs.openstack.org/27/503627/2/check/gate-instack-undercloud-requirements/3f0f66f/ : SUCCESS in 1m 28s\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/27/503627/2/check/gate-instack-undercloud-releasenotes/a934e74//releasenotes/build/html/ : SUCCESS in 1m 52s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/27/503627/2/check/gate-instack-undercloud-puppet-syntax-3-centos-7/2294ede/ : SUCCESS in 4m 31s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/27/503627/2/check/gate-instack-undercloud-puppet-syntax-4-centos-7/2bab133/ : SUCCESS in 4m 27s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/27/503627/2/check/gate-instack-undercloud-puppet-lint/d822d28/ : SUCCESS in 2m 24s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/27/503627/2/check/instack-undercloud-coverage-ubuntu-xenial-nv/3ca227f/ : SUCCESS in 2m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9503c0f5ea5dd158b5b2cd81bcff16971b1fa97f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2017-09-18 09:50:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"17ea191226dba759b14812bb63f79acc09a8aec6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 01:07:30.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"e9fb10c980e30e9eef0280e20220f1fd6d753be4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-20 03:01:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/27/503627/3/check/gate-instack-undercloud-docs-ubuntu-xenial/30421b9//doc/build/html/ : SUCCESS in 2m 54s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/27/503627/3/check/gate-instack-undercloud-pep8-ubuntu-xenial/86a64b5/ : SUCCESS in 2m 11s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/27/503627/3/check/gate-instack-undercloud-python27-ubuntu-xenial/71f84ad/ : SUCCESS in 3m 38s\n- gate-instack-undercloud-python35 http://logs.openstack.org/27/503627/3/check/gate-instack-undercloud-python35/6a2b90a/ : SUCCESS in 3m 24s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/27/503627/3/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/f8806b4/ : SUCCESS in 1h 49m 46s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/27/503627/3/check/gate-tripleo-ci-centos-7-containers-multinode/401d278/ : SUCCESS in 1h 43m 52s\n- gate-tripleo-ci-centos-7-undercloud-containers-nv http://logs.openstack.org/27/503627/3/check/gate-tripleo-ci-centos-7-undercloud-containers-nv/4df45a1/ : FAILURE in 44m 34s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/27/503627/3/check/gate-tripleo-ci-centos-7-undercloud-oooq/35db1bb/ : SUCCESS in 53m 35s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/27/503627/3/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/7492001/ : SUCCESS in 1h 05m 54s (non-voting)\n- gate-instack-undercloud-requirements http://logs.openstack.org/27/503627/3/check/gate-instack-undercloud-requirements/ea3f68e/ : SUCCESS in 1m 01s\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/27/503627/3/check/gate-instack-undercloud-releasenotes/f39ef3d//releasenotes/build/html/ : SUCCESS in 1m 54s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/27/503627/3/check/gate-instack-undercloud-puppet-syntax-3-centos-7/e6cb05e/ : SUCCESS in 3m 47s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/27/503627/3/check/gate-instack-undercloud-puppet-syntax-4-centos-7/a07e964/ : SUCCESS in 4m 05s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/27/503627/3/check/gate-instack-undercloud-puppet-lint/5248cd5/ : SUCCESS in 3m 13s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/27/503627/3/check/instack-undercloud-coverage-ubuntu-xenial-nv/b32c1ee/ : SUCCESS in 2m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"138a30e1ccd2dcc6f1733a982ede970ab664b3e8","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 03:34:14.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"fb3bb2a25d4a4dd7c28e4ab3a88a70162f41ca91","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-20 06:11:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/27/503627/4/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/953c8ee/ : SUCCESS in 2h 37m 29s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/27/503627/4/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq/0761a2f/ : SUCCESS in 2h 34m 02s\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/27/503627/4/check-tripleo/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/2120805/ : SUCCESS in 2h 36m 53s","accounts_in_message":[],"_revision_number":4},{"id":"4f35fcd28297e18b959270d1d27aeaf38cf85a56","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-20 06:21:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/27/503627/4/check/gate-instack-undercloud-docs-ubuntu-xenial/9b1164d//doc/build/html/ : SUCCESS in 2m 43s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/27/503627/4/check/gate-instack-undercloud-pep8-ubuntu-xenial/3478cb9/ : SUCCESS in 2m 05s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/27/503627/4/check/gate-instack-undercloud-python27-ubuntu-xenial/0a76481/ : SUCCESS in 2m 06s\n- gate-instack-undercloud-python35 http://logs.openstack.org/27/503627/4/check/gate-instack-undercloud-python35/9048a12/ : SUCCESS in 2m 33s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/27/503627/4/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/d4f57ca/ : FAILURE in 2h 40m 24s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/27/503627/4/check/gate-tripleo-ci-centos-7-containers-multinode/dea757d/ : FAILURE in 2h 42m 35s\n- gate-tripleo-ci-centos-7-undercloud-containers-nv http://logs.openstack.org/27/503627/4/check/gate-tripleo-ci-centos-7-undercloud-containers-nv/8c4e318/ : FAILURE in 52m 39s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/27/503627/4/check/gate-tripleo-ci-centos-7-undercloud-oooq/97e12fb/ : SUCCESS in 1h 05m 46s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/27/503627/4/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/df6f17a/ : SUCCESS in 1h 07m 11s (non-voting)\n- gate-instack-undercloud-requirements http://logs.openstack.org/27/503627/4/check/gate-instack-undercloud-requirements/c49cf6e/ : SUCCESS in 1m 09s\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/27/503627/4/check/gate-instack-undercloud-releasenotes/6217585//releasenotes/build/html/ : SUCCESS in 1m 50s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/27/503627/4/check/gate-instack-undercloud-puppet-syntax-3-centos-7/ad2dcf8/ : SUCCESS in 3m 37s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/27/503627/4/check/gate-instack-undercloud-puppet-syntax-4-centos-7/27ce459/ : SUCCESS in 3m 01s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/27/503627/4/check/gate-instack-undercloud-puppet-lint/899e2d8/ : SUCCESS in 2m 07s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/27/503627/4/check/instack-undercloud-coverage-ubuntu-xenial-nv/473f1e5/ : SUCCESS in 1m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b56a36f124374ffff87570ac56b46a748b749fcc","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-09-20 11:34:51.000000000","message":"Patch Set 4:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tripleo-ci-centos-7-containers-multinode: https://bugs.launchpad.net/bugs/1686542\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq: https://bugs.launchpad.net/bugs/1686542\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":4},{"id":"4398b1abe4193e24ccf5c21ca9b79d3b94b50caa","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 13:49:16.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"861ce714b6dfe8db7a41f1300800fa5aa318f99c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 14:10:47.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"1cd34c2064eb8fe2ba2a4ef80ecf6df625062d2c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 14:33:02.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"b29f01ef34f83091f56a277323273bee6f51f424","author":{"_account_id":9317,"name":"Ana Krivokapic","email":"akrivoka@redhat.com","username":"akrivoka"},"date":"2017-09-20 14:40:14.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"569280172b159edbffaadd78eb3486ccedd2c142","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 14:50:15.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"d9a0e0f44232d701c48908d52c8c598f8bd79a5d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 15:07:10.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"008ffbe7ed24f3f5475dfe305aa430e94592bdc2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 15:24:21.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"f88b58b1a42d3f483a9468e7154e101522c35969","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 15:41:09.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"8eeb8b388f4bbbb14dfe8fd8efeed122c8169236","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 15:57:56.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"dcfafdccabbd0eaf797ecbac9136d589ceeacb13","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 16:14:47.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"1785f97f431a64878d1aace3823cf8dc8754ab1f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-20 17:58:31.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/27/503627/13/check/gate-instack-undercloud-docs-ubuntu-xenial/b39640e//doc/build/html/ : SUCCESS in 2m 25s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/27/503627/13/check/gate-instack-undercloud-pep8-ubuntu-xenial/5d1b72a/ : SUCCESS in 4m 11s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/27/503627/13/check/gate-instack-undercloud-python27-ubuntu-xenial/20ee51a/ : SUCCESS in 4m 22s\n- gate-instack-undercloud-python35 http://logs.openstack.org/27/503627/13/check/gate-instack-undercloud-python35/a054671/ : SUCCESS in 4m 56s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/27/503627/13/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/8b51200/ : SUCCESS in 1h 39m 01s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/27/503627/13/check/gate-tripleo-ci-centos-7-containers-multinode/0ad54a9/ : SUCCESS in 1h 37m 24s\n- gate-tripleo-ci-centos-7-undercloud-containers-nv http://logs.openstack.org/27/503627/13/check/gate-tripleo-ci-centos-7-undercloud-containers-nv/38a52cf/ : FAILURE in 41m 16s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/27/503627/13/check/gate-tripleo-ci-centos-7-undercloud-oooq/976d471/ : SUCCESS in 1h 00m 13s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/27/503627/13/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/e8ff398/ : SUCCESS in 1h 03m 50s (non-voting)\n- gate-instack-undercloud-requirements http://logs.openstack.org/27/503627/13/check/gate-instack-undercloud-requirements/ae89c6e/ : SUCCESS in 3m 36s\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/27/503627/13/check/gate-instack-undercloud-releasenotes/b032d6d//releasenotes/build/html/ : SUCCESS in 3m 36s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/27/503627/13/check/gate-instack-undercloud-puppet-syntax-3-centos-7/e46edf2/ : SUCCESS in 4m 38s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/27/503627/13/check/gate-instack-undercloud-puppet-syntax-4-centos-7/c679226/ : SUCCESS in 3m 52s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/27/503627/13/check/gate-instack-undercloud-puppet-lint/599ed1c/ : SUCCESS in 4m 09s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/27/503627/13/check/instack-undercloud-coverage-ubuntu-xenial-nv/6d10400/ : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b24e80ac7f090d935de7c7d7ced082e4b952976c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-20 19:18:36.000000000","message":"Patch Set 13:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/27/503627/13/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/69fae90/ : SUCCESS in 2h 45m 36s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/27/503627/13/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq/b776128/ : SUCCESS in 2h 46m 09s\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/27/503627/13/check-tripleo/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/1b5801b/ : SUCCESS in 2h 46m 02s","accounts_in_message":[],"_revision_number":13},{"id":"aadfd3cf20790f27757ca7eb66c3531ec7ebdfad","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 21:53:40.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"3c1795b13a04cf7860717276ed8336119c382fd4","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-20 22:20:11.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"e5c4b2913b06ef18457a3703eabe762ab24b7370","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-21 00:22:04.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/27/503627/15/check/gate-instack-undercloud-docs-ubuntu-xenial/9445e86//doc/build/html/ : SUCCESS in 2m 09s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/27/503627/15/check/gate-instack-undercloud-pep8-ubuntu-xenial/c2d0643/ : SUCCESS in 3m 01s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/27/503627/15/check/gate-instack-undercloud-python27-ubuntu-xenial/3bddcea/ : SUCCESS in 3m 06s\n- gate-instack-undercloud-python35 http://logs.openstack.org/27/503627/15/check/gate-instack-undercloud-python35/00f47f0/ : SUCCESS in 2m 07s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/27/503627/15/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/b329c02/ : SUCCESS in 1h 47m 23s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/27/503627/15/check/gate-tripleo-ci-centos-7-containers-multinode/4d9736e/ : SUCCESS in 1h 36m 25s\n- gate-tripleo-ci-centos-7-undercloud-containers-nv http://logs.openstack.org/27/503627/15/check/gate-tripleo-ci-centos-7-undercloud-containers-nv/1bb70cf/ : SUCCESS in 41m 44s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/27/503627/15/check/gate-tripleo-ci-centos-7-undercloud-oooq/a2ce5d5/ : SUCCESS in 1h 02m 06s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/27/503627/15/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/133d740/ : SUCCESS in 1h 12m 08s (non-voting)\n- gate-instack-undercloud-requirements http://logs.openstack.org/27/503627/15/check/gate-instack-undercloud-requirements/5948c47/ : SUCCESS in 2m 42s\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/27/503627/15/check/gate-instack-undercloud-releasenotes/68ea5ff//releasenotes/build/html/ : SUCCESS in 3m 04s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/27/503627/15/check/gate-instack-undercloud-puppet-syntax-3-centos-7/9913a93/ : SUCCESS in 4m 33s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/27/503627/15/check/gate-instack-undercloud-puppet-syntax-4-centos-7/2daf4f4/ : SUCCESS in 4m 38s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/27/503627/15/check/gate-instack-undercloud-puppet-lint/aa22319/ : SUCCESS in 3m 14s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/27/503627/15/check/instack-undercloud-coverage-ubuntu-xenial-nv/a037096/ : SUCCESS in 2m 36s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"11841a6d804e9caf113d09a040e1ca9adf28516d","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-21 00:33:45.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"51f308b5f068d279c40da44d50dfece9d62342d6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-21 01:12:22.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"e2a63bfa4d80d79cbb3ada9a64756df6bc13a8ff","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-21 02:06:08.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"f1996b8657ae639af71c9aa162bf69dd21eac4be","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-21 03:05:13.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.","accounts_in_message":[],"_revision_number":19},{"id":"462f3277cf439ca0b159fa9a4ed13a32f2de914c","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-21 03:22:05.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"ca8d4cd16d9e91a32892053882a43a91009158b0","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-21 03:40:30.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"58152dedaca767e0b8b88deb1e50839c2e637b68","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-21 05:42:25.000000000","message":"Patch Set 21:\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/27/503627/21/check/gate-instack-undercloud-docs-ubuntu-xenial/c8a26fa//doc/build/html/ : SUCCESS in 1m 58s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/27/503627/21/check/gate-instack-undercloud-pep8-ubuntu-xenial/86969bd/ : SUCCESS in 3m 00s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/27/503627/21/check/gate-instack-undercloud-python27-ubuntu-xenial/d64e452/ : SUCCESS in 2m 29s\n- gate-instack-undercloud-python35 http://logs.openstack.org/27/503627/21/check/gate-instack-undercloud-python35/c438bf8/ : SUCCESS in 2m 29s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/27/503627/21/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/4b9c9a4/ : SUCCESS in 1h 56m 20s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/27/503627/21/check/gate-tripleo-ci-centos-7-containers-multinode/44d6d32/ : SUCCESS in 1h 28m 46s\n- gate-tripleo-ci-centos-7-undercloud-containers-nv http://logs.openstack.org/27/503627/21/check/gate-tripleo-ci-centos-7-undercloud-containers-nv/291b3ea/ : SUCCESS in 45m 52s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/27/503627/21/check/gate-tripleo-ci-centos-7-undercloud-oooq/f3043cb/ : SUCCESS in 46m 47s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/27/503627/21/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/c994075/ : SUCCESS in 1h 01m 29s (non-voting)\n- gate-instack-undercloud-requirements http://logs.openstack.org/27/503627/21/check/gate-instack-undercloud-requirements/cb06a57/ : SUCCESS in 1m 22s\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/27/503627/21/check/gate-instack-undercloud-releasenotes/2923760//releasenotes/build/html/ : SUCCESS in 2m 49s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/27/503627/21/check/gate-instack-undercloud-puppet-syntax-3-centos-7/245401f/ : SUCCESS in 3m 53s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/27/503627/21/check/gate-instack-undercloud-puppet-syntax-4-centos-7/92734b4/ : SUCCESS in 3m 06s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/27/503627/21/check/gate-instack-undercloud-puppet-lint/d6ab89d/ : SUCCESS in 2m 23s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/27/503627/21/check/instack-undercloud-coverage-ubuntu-xenial-nv/4f77cea/ : SUCCESS in 1m 55s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"841535e44a329ac5bdea04d5eae4be324df1162f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-21 06:18:00.000000000","message":"Patch Set 21:\n\nBuild succeeded (check-tripleo pipeline).\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/27/503627/21/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/0b13706/ : SUCCESS in 2h 37m 00s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/27/503627/21/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq/568e3ab/ : SUCCESS in 2h 31m 32s\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/27/503627/21/check-tripleo/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/5e8d77b/ : SUCCESS in 2h 36m 10s","accounts_in_message":[],"_revision_number":21},{"id":"92b9a8d91213076ebcd14e1ff8eaef93c56af649","author":{"_account_id":25747,"name":"Debo Zhang","email":"zhangdebo@inspur.com","username":"zhangdebo"},"date":"2017-09-21 06:26:18.000000000","message":"Patch Set 21: Code-Review+1","accounts_in_message":[],"_revision_number":21},{"id":"c9c129ce33de4851ef0856902bd3fe6da9a7e15f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-21 12:45:57.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"ae21e2f124cd43ed25b1f5a5b850f18f189657e2","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-21 13:01:47.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"1fb0b190b92fe1c6eb3ee4668ff4888de52cffeb","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-21 13:29:13.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"8b65e0f169f9522c67a950b0e8f467b3c0aa1fe5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-21 14:39:14.000000000","message":"Patch Set 24:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/27/503627/24/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/a258ffd/ : FAILURE in 1h 04m 59s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/27/503627/24/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq/0111913/ : FAILURE in 1h 07m 54s\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/27/503627/24/check-tripleo/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/1a13009/ : FAILURE in 1h 09m 23s","accounts_in_message":[],"_revision_number":24},{"id":"8ad48ee7fafc1a97a9ad0df4e4acea7485f0c1b9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-21 15:18:31.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/27/503627/24/check/gate-instack-undercloud-docs-ubuntu-xenial/63b1de8//doc/build/html/ : SUCCESS in 2m 53s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/27/503627/24/check/gate-instack-undercloud-pep8-ubuntu-xenial/d6ad4e2/ : SUCCESS in 2m 54s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/27/503627/24/check/gate-instack-undercloud-python27-ubuntu-xenial/21ca48d/ : SUCCESS in 3m 01s\n- gate-instack-undercloud-python35 http://logs.openstack.org/27/503627/24/check/gate-instack-undercloud-python35/4940c7a/ : SUCCESS in 3m 28s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/27/503627/24/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/28de52f/ : SUCCESS in 1h 48m 40s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/27/503627/24/check/gate-tripleo-ci-centos-7-containers-multinode/85da1b8/ : SUCCESS in 1h 28m 20s\n- gate-tripleo-ci-centos-7-undercloud-containers-nv http://logs.openstack.org/27/503627/24/check/gate-tripleo-ci-centos-7-undercloud-containers-nv/b3b4e90/ : FAILURE in 22m 04s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/27/503627/24/check/gate-tripleo-ci-centos-7-undercloud-oooq/c62720d/ : SUCCESS in 56m 36s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/27/503627/24/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/ba17f58/ : SUCCESS in 59m 11s (non-voting)\n- gate-instack-undercloud-requirements http://logs.openstack.org/27/503627/24/check/gate-instack-undercloud-requirements/d895de4/ : SUCCESS in 1m 52s\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/27/503627/24/check/gate-instack-undercloud-releasenotes/6a5e75f//releasenotes/build/html/ : SUCCESS in 2m 52s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/27/503627/24/check/gate-instack-undercloud-puppet-syntax-3-centos-7/95fee6f/ : SUCCESS in 3m 05s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/27/503627/24/check/gate-instack-undercloud-puppet-syntax-4-centos-7/f2c724b/ : SUCCESS in 3m 01s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/27/503627/24/check/gate-instack-undercloud-puppet-lint/bb2e543/ : SUCCESS in 3m 38s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/27/503627/24/check/instack-undercloud-coverage-ubuntu-xenial-nv/abb06fa/ : SUCCESS in 3m 26s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"833d17601d45fd821794546cd72273251997a666","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-21 20:51:23.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"7ae4eaa70fe7ea671900aab5da57d4c7b4cc0817","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-09-21 20:53:14.000000000","message":"Patch Set 24:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024: unrecognized error\n- gate-tripleo-ci-centos-7-ovb-containers-oooq: unrecognized error\n- gate-tripleo-ci-centos-7-ovb-ha-oooq: https://bugs.launchpad.net/bugs/1708704\n\nSome of the tests failed in a way that we did not understand. Please help us classify these issues so that they can be part of Elastic Recheck http://status.openstack.org/elastic-recheck/\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":24},{"id":"869fee66ecf1cb5a00610cd8f28ce1de5bc18e53","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-21 21:08:21.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"75607ce0cec25501368b87e2bee4aa995ac1ab4b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-21 21:35:53.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"358b77ab4eb5140c98e0fe9ffa05338656140fd1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-21 22:45:04.000000000","message":"Patch Set 27:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/27/503627/27/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/6bfffa4/ : FAILURE in 1h 08m 54s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/27/503627/27/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq/e112a43/ : FAILURE in 1h 04m 49s\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/27/503627/27/check-tripleo/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/88899ea/ : FAILURE in 1h 07m 45s","accounts_in_message":[],"_revision_number":27},{"id":"2d07bde746315f658e5dc98157db5b12fe816a11","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-21 23:40:01.000000000","message":"Patch Set 27:\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/27/503627/27/check/gate-instack-undercloud-docs-ubuntu-xenial/584ceea//doc/build/html/ : SUCCESS in 2m 31s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/27/503627/27/check/gate-instack-undercloud-pep8-ubuntu-xenial/c8d889c/ : SUCCESS in 2m 51s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/27/503627/27/check/gate-instack-undercloud-python27-ubuntu-xenial/46bae57/ : SUCCESS in 2m 51s\n- gate-instack-undercloud-python35 http://logs.openstack.org/27/503627/27/check/gate-instack-undercloud-python35/4508398/ : SUCCESS in 2m 52s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/27/503627/27/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/ef903d4/ : SUCCESS in 1h 59m 41s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/27/503627/27/check/gate-tripleo-ci-centos-7-containers-multinode/8ff338b/ : SUCCESS in 1h 47m 11s\n- gate-tripleo-ci-centos-7-undercloud-containers-nv http://logs.openstack.org/27/503627/27/check/gate-tripleo-ci-centos-7-undercloud-containers-nv/d8f09d5/ : SUCCESS in 1h 01m 40s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/27/503627/27/check/gate-tripleo-ci-centos-7-undercloud-oooq/a39ec81/ : SUCCESS in 1h 10m 59s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/27/503627/27/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/9fe8701/ : SUCCESS in 1h 14m 18s (non-voting)\n- gate-instack-undercloud-requirements http://logs.openstack.org/27/503627/27/check/gate-instack-undercloud-requirements/ddc7947/ : SUCCESS in 1m 07s\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/27/503627/27/check/gate-instack-undercloud-releasenotes/4af85ed//releasenotes/build/html/ : SUCCESS in 2m 02s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/27/503627/27/check/gate-instack-undercloud-puppet-syntax-3-centos-7/ee09f7c/ : SUCCESS in 8m 26s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/27/503627/27/check/gate-instack-undercloud-puppet-syntax-4-centos-7/8c3575f/ : SUCCESS in 8m 42s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/27/503627/27/check/gate-instack-undercloud-puppet-lint/21e8adb/ : SUCCESS in 2m 30s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/27/503627/27/check/instack-undercloud-coverage-ubuntu-xenial-nv/f56c286/ : SUCCESS in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"f055bb43982af342b3c8a95b08a4023e5918396b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2017-09-22 12:51:36.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"0e95548c73e7be133336e0c69c9516b78625c55e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2017-09-22 13:30:43.000000000","message":"Patch Set 28: Workflow+1 Code-Review+2","accounts_in_message":[],"_revision_number":28},{"id":"a2d47b24023484ed804bb1821b954b9caffba6d1","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2017-09-22 13:31:18.000000000","message":"Patch Set 28:\n\nI should note that the requirements files aren\u0027t used in OVB jobs, so as long as this passes the gate jobs it\u0027s fine.","accounts_in_message":[],"_revision_number":28},{"id":"7e12f04c0aac4a228ac7f669ee86f247ca5f869f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 15:25:41.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/27/503627/28/check/gate-instack-undercloud-docs-ubuntu-xenial/df82c3f//doc/build/html/ : SUCCESS in 3m 07s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/27/503627/28/check/gate-instack-undercloud-pep8-ubuntu-xenial/78a7272/ : SUCCESS in 2m 07s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/27/503627/28/check/gate-instack-undercloud-python27-ubuntu-xenial/72b7385/ : SUCCESS in 2m 49s\n- gate-instack-undercloud-python35 http://logs.openstack.org/27/503627/28/check/gate-instack-undercloud-python35/102554f/ : SUCCESS in 4m 16s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/27/503627/28/check/gate-tripleo-ci-centos-7-nonha-multinode-oooq/287e8e2/ : SUCCESS in 2h 28m 35s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/27/503627/28/check/gate-tripleo-ci-centos-7-containers-multinode/afcbf4a/ : SUCCESS in 2h 10m 19s\n- gate-tripleo-ci-centos-7-undercloud-containers-nv http://logs.openstack.org/27/503627/28/check/gate-tripleo-ci-centos-7-undercloud-containers-nv/263d5b4/ : SUCCESS in 1h 23m 52s (non-voting)\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/27/503627/28/check/gate-tripleo-ci-centos-7-undercloud-oooq/8b22194/ : SUCCESS in 1h 28m 53s\n- gate-tripleo-ci-centos-7-undercloud-upgrades-nv http://logs.openstack.org/27/503627/28/check/gate-tripleo-ci-centos-7-undercloud-upgrades-nv/cb2bba0/ : FAILURE in 1h 34m 19s (non-voting)\n- gate-instack-undercloud-requirements http://logs.openstack.org/27/503627/28/check/gate-instack-undercloud-requirements/d4f4325/ : SUCCESS in 1m 38s\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/27/503627/28/check/gate-instack-undercloud-releasenotes/2b23100//releasenotes/build/html/ : SUCCESS in 3m 18s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/27/503627/28/check/gate-instack-undercloud-puppet-syntax-3-centos-7/7ec00ef/ : SUCCESS in 4m 35s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/27/503627/28/check/gate-instack-undercloud-puppet-syntax-4-centos-7/b81cdf5/ : SUCCESS in 4m 29s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/27/503627/28/check/gate-instack-undercloud-puppet-lint/9390c38/ : SUCCESS in 2m 31s\n- instack-undercloud-coverage-ubuntu-xenial-nv http://logs.openstack.org/27/503627/28/check/instack-undercloud-coverage-ubuntu-xenial-nv/dcb14f7/ : SUCCESS in 2m 48s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"5edaa413b39b9e15251d51e6bc1d383ffd631ea5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 15:26:01.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":28},{"id":"2f5fe2db4861c540917ddb014ca21728504febd2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 15:39:03.000000000","message":"Patch Set 28:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/27/503627/28/check-tripleo/gate-tripleo-ci-centos-7-ovb-ha-oooq/b5c8df7/ : FAILURE in 2h 47m 10s\n- gate-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/27/503627/28/check-tripleo/gate-tripleo-ci-centos-7-ovb-containers-oooq/0a87d70/ : SUCCESS in 2h 34m 19s\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/27/503627/28/check-tripleo/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/d2766f3/ : SUCCESS in 2h 33m 38s","accounts_in_message":[],"_revision_number":28},{"id":"50e9ef20f5534868e2463ee4a8195f04de32c988","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2017-09-22 15:55:11.000000000","message":"Patch Set 28:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-tripleo-ci-centos-7-ovb-ha-oooq: https://bugs.launchpad.net/bugs/1686542\n\nIf you believe we\u0027ve correctly identified the failure, feel free to leave a \u0027recheck\u0027 comment to run the tests again.\nFor more details on this and other bugs, please see http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":28},{"id":"d9b815e2e5f8f91793f7b1ef3063e561b239f72b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 19:53:52.000000000","message":"Patch Set 28: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-instack-undercloud-docs-ubuntu-xenial http://docs-draft.openstack.org/27/503627/28/gate/gate-instack-undercloud-docs-ubuntu-xenial/cf94ea0//doc/build/html/ : SUCCESS in 2m 29s\n- gate-instack-undercloud-pep8-ubuntu-xenial http://logs.openstack.org/27/503627/28/gate/gate-instack-undercloud-pep8-ubuntu-xenial/7e45868/ : SUCCESS in 2m 16s\n- gate-instack-undercloud-python27-ubuntu-xenial http://logs.openstack.org/27/503627/28/gate/gate-instack-undercloud-python27-ubuntu-xenial/2bcd82b/ : SUCCESS in 3m 19s\n- gate-instack-undercloud-python35 http://logs.openstack.org/27/503627/28/gate/gate-instack-undercloud-python35/01b9c8e/ : SUCCESS in 2m 47s\n- gate-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/27/503627/28/gate/gate-tripleo-ci-centos-7-nonha-multinode-oooq/74eb705/ : SUCCESS in 1h 54m 42s\n- gate-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/27/503627/28/gate/gate-tripleo-ci-centos-7-containers-multinode/8666683/ : SUCCESS in 1h 45m 17s\n- gate-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/27/503627/28/gate/gate-tripleo-ci-centos-7-undercloud-oooq/2cf2db1/ : SUCCESS in 57m 23s\n- gate-instack-undercloud-requirements http://logs.openstack.org/27/503627/28/gate/gate-instack-undercloud-requirements/8e79fdf/ : SUCCESS in 1m 15s\n- gate-instack-undercloud-releasenotes http://docs-draft.openstack.org/27/503627/28/gate/gate-instack-undercloud-releasenotes/1d96f8c//releasenotes/build/html/ : SUCCESS in 2m 29s\n- gate-instack-undercloud-puppet-syntax-3-centos-7 http://logs.openstack.org/27/503627/28/gate/gate-instack-undercloud-puppet-syntax-3-centos-7/ef5c14e/ : SUCCESS in 4m 40s\n- gate-instack-undercloud-puppet-syntax-4-centos-7 http://logs.openstack.org/27/503627/28/gate/gate-instack-undercloud-puppet-syntax-4-centos-7/53ff4ea/ : SUCCESS in 4m 04s\n- gate-instack-undercloud-puppet-lint http://logs.openstack.org/27/503627/28/gate/gate-instack-undercloud-puppet-lint/22e440d/ : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":28},{"id":"8340c2ae17fd62566d7e6dda087e255b9d874b0f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-09-22 19:53:53.000000000","message":"Change has been successfully merged by Jenkins","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"fc1c3ee3029b5175e329a6ec7639cec2f2d994c9","revisions":{"4f47de3b2a1bf58fd227e52f858c272355bdb6e2":{"kind":"REWORK","_number":1,"created":"2017-09-13 12:51:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/1"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-13 12:51:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-13 12:51:10.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/4f47de3b2a1bf58fd227e52f858c272355bdb6e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/4f47de3b2a1bf58fd227e52f858c272355bdb6e2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6304be2bcc5fd3f5885a48e3cf512ea43a8703f5":{"kind":"REWORK","_number":2,"created":"2017-09-16 23:12:49.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/2"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-16 23:12:49.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-16 23:12:49.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/6304be2bcc5fd3f5885a48e3cf512ea43a8703f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/6304be2bcc5fd3f5885a48e3cf512ea43a8703f5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"bbeeb27afdb9e6dacc0d1827a98fefb46708732e":{"kind":"NO_CHANGE","_number":3,"created":"2017-09-20 01:07:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/3"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 01:07:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 01:07:28.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/bbeeb27afdb9e6dacc0d1827a98fefb46708732e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/bbeeb27afdb9e6dacc0d1827a98fefb46708732e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"114007c0a3a6e6b33ea04f1f350480de07d090f9":{"kind":"NO_CHANGE","_number":4,"created":"2017-09-20 03:34:14.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/4"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 03:34:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 03:34:13.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/114007c0a3a6e6b33ea04f1f350480de07d090f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/114007c0a3a6e6b33ea04f1f350480de07d090f9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"24be48c5bde22054f8dacf9066f3a62854b35a36":{"kind":"NO_CHANGE","_number":5,"created":"2017-09-20 13:49:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/5"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 13:49:15.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 13:49:15.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/24be48c5bde22054f8dacf9066f3a62854b35a36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/24be48c5bde22054f8dacf9066f3a62854b35a36"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"16efa93689a3d609638f1ef36e42c97528aa7b97":{"kind":"NO_CHANGE","_number":6,"created":"2017-09-20 14:10:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/6"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 14:10:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 14:10:46.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/16efa93689a3d609638f1ef36e42c97528aa7b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/16efa93689a3d609638f1ef36e42c97528aa7b97"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3077b3974597ea992f0b98bea5b6ef013d494f3e":{"kind":"NO_CHANGE","_number":7,"created":"2017-09-20 14:33:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/7"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 14:33:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 14:33:01.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/3077b3974597ea992f0b98bea5b6ef013d494f3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/3077b3974597ea992f0b98bea5b6ef013d494f3e"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"097ac8c35be790e033add448858f5b28e2f72c6a":{"kind":"NO_CHANGE","_number":8,"created":"2017-09-20 14:50:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/8"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 14:50:14.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 14:50:14.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/097ac8c35be790e033add448858f5b28e2f72c6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/097ac8c35be790e033add448858f5b28e2f72c6a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a7deb822382abc196627e9644f3044a011ff5478":{"kind":"NO_CHANGE","_number":9,"created":"2017-09-20 15:07:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/9"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 15:07:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 15:07:10.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/a7deb822382abc196627e9644f3044a011ff5478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/a7deb822382abc196627e9644f3044a011ff5478"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0e7e77c36aba31ca7ffcbd5401bf89dfe25c8f95":{"kind":"NO_CHANGE","_number":10,"created":"2017-09-20 15:24:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/10"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 15:24:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 15:24:20.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/0e7e77c36aba31ca7ffcbd5401bf89dfe25c8f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/0e7e77c36aba31ca7ffcbd5401bf89dfe25c8f95"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"1dd82179734359b2096720109121a9cb316fa025":{"kind":"NO_CHANGE","_number":11,"created":"2017-09-20 15:41:09.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/11"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 15:41:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 15:41:08.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1dd82179734359b2096720109121a9cb316fa025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1dd82179734359b2096720109121a9cb316fa025"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b988fcf4a56c7d0e357beb6d7769a6e6c508f52f":{"kind":"NO_CHANGE","_number":12,"created":"2017-09-20 15:57:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/12"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 15:57:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 15:57:56.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/b988fcf4a56c7d0e357beb6d7769a6e6c508f52f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/b988fcf4a56c7d0e357beb6d7769a6e6c508f52f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c002be74d3ff9edc300c2946ecb12510ba0a0a83":{"kind":"NO_CHANGE","_number":13,"created":"2017-09-20 16:14:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/13"}}},"commit":{"parents":[{"commit":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","subject":"Updated from global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/1570a9cf0f7167e794c0b366f2dfdc1eea3e7912"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 16:14:46.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 16:14:46.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/c002be74d3ff9edc300c2946ecb12510ba0a0a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/c002be74d3ff9edc300c2946ecb12510ba0a0a83"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1570a9cf0f7167e794c0b366f2dfdc1eea3e7912","is_merged_in_target_branch":true,"change_id":"Ifde981b6a04ffd35ec8f189eaaafaf999bedd21f","change_number":502651,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4080a914c6ea0e431633ec896cb99edfae1bdec2":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2017-09-20 21:53:40.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/14"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 21:53:39.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 21:53:39.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/4080a914c6ea0e431633ec896cb99edfae1bdec2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/4080a914c6ea0e431633ec896cb99edfae1bdec2"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"dc7ba2173bab87d13ceed79e6b07e495a6f149de":{"kind":"NO_CHANGE","_number":15,"created":"2017-09-20 22:20:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/15"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 22:20:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-20 22:20:10.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/dc7ba2173bab87d13ceed79e6b07e495a6f149de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/dc7ba2173bab87d13ceed79e6b07e495a6f149de"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"a558244664c01304a99081c0d6215f59b80ef7b1":{"kind":"NO_CHANGE","_number":16,"created":"2017-09-21 00:33:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/16"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 00:33:43.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 00:33:43.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/a558244664c01304a99081c0d6215f59b80ef7b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/a558244664c01304a99081c0d6215f59b80ef7b1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c0142fd4ce4a4580e967368d8cda74ca3143c118":{"kind":"NO_CHANGE","_number":17,"created":"2017-09-21 01:12:22.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/17"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 01:12:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 01:12:21.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/c0142fd4ce4a4580e967368d8cda74ca3143c118"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/c0142fd4ce4a4580e967368d8cda74ca3143c118"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"3b5c874b7f526411315c2025e67b01adb33717aa":{"kind":"NO_CHANGE","_number":18,"created":"2017-09-21 02:06:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/18"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 02:06:07.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 02:06:07.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/3b5c874b7f526411315c2025e67b01adb33717aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/3b5c874b7f526411315c2025e67b01adb33717aa"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7b642d23e616a72d246268b1f8ad676b4844368c":{"kind":"NO_CHANGE","_number":19,"created":"2017-09-21 03:05:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/19"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 03:05:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 03:05:12.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/7b642d23e616a72d246268b1f8ad676b4844368c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/7b642d23e616a72d246268b1f8ad676b4844368c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"18c653803eb536abe05cd5e48006b81453a6a9bc":{"kind":"NO_CHANGE","_number":20,"created":"2017-09-21 03:22:05.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/20"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 03:22:04.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 03:22:04.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/18c653803eb536abe05cd5e48006b81453a6a9bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/18c653803eb536abe05cd5e48006b81453a6a9bc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b5c83dcff279e0a179967045003dc71299713303":{"kind":"NO_CHANGE","_number":21,"created":"2017-09-21 03:40:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/21"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 03:40:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 03:40:29.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/b5c83dcff279e0a179967045003dc71299713303"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/b5c83dcff279e0a179967045003dc71299713303"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"626b44a506fcc14adfb9b3121b1d8c6ecb47a8c5":{"kind":"NO_CHANGE","_number":22,"created":"2017-09-21 12:45:57.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/22"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 12:45:56.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 12:45:56.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/626b44a506fcc14adfb9b3121b1d8c6ecb47a8c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/626b44a506fcc14adfb9b3121b1d8c6ecb47a8c5"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c6f4e5831edab144f4c453a664cf4031a5d4cc53":{"kind":"NO_CHANGE","_number":23,"created":"2017-09-21 13:01:47.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/23"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 13:01:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 13:01:47.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/c6f4e5831edab144f4c453a664cf4031a5d4cc53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/c6f4e5831edab144f4c453a664cf4031a5d4cc53"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"cfb0a75c8531491d931cd61a5564726fc2ed8fae":{"kind":"NO_CHANGE","_number":24,"created":"2017-09-21 13:29:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/24"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 13:29:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 13:29:12.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/cfb0a75c8531491d931cd61a5564726fc2ed8fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/cfb0a75c8531491d931cd61a5564726fc2ed8fae"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"b67a39a00d3611588b88013c551db5306a191705":{"kind":"NO_CHANGE","_number":25,"created":"2017-09-21 20:51:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/25"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 20:51:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 20:51:22.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/b67a39a00d3611588b88013c551db5306a191705"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/b67a39a00d3611588b88013c551db5306a191705"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"baa18d8bb7613eb8fe5ba390ca95402a1d454441":{"kind":"NO_CHANGE","_number":26,"created":"2017-09-21 21:08:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/26"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 21:08:21.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 21:08:21.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/baa18d8bb7613eb8fe5ba390ca95402a1d454441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/baa18d8bb7613eb8fe5ba390ca95402a1d454441"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"300c0f189d34296baeee0ab44d370b26fd1e3693":{"kind":"NO_CHANGE","_number":27,"created":"2017-09-21 21:35:53.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/27"}}},"commit":{"parents":[{"commit":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","subject":"Increase the size of the Mistral output limit","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/83bf89ee590e13afc88281fc0ef52e45a77f7cb4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 21:35:52.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-21 21:35:52.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/300c0f189d34296baeee0ab44d370b26fd1e3693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/300c0f189d34296baeee0ab44d370b26fd1e3693"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"83bf89ee590e13afc88281fc0ef52e45a77f7cb4","is_merged_in_target_branch":true,"change_id":"Ief9e8402f0251d1447a825f496050334a8d80dc5","change_number":505582,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"fc1c3ee3029b5175e329a6ec7639cec2f2d994c9":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2017-09-22 12:51:36.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/503627/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/27/503627/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/27/503627/28"}}},"commit":{"parents":[{"commit":"674d304ee7865138be23f7183aede125e3526496","subject":"Merge \"Deploy Mistral with Keystone v3 options (authtoken)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/674d304ee7865138be23f7183aede125e3526496"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-22 12:51:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2017-09-22 12:51:35.000000000","tz":0},"subject":"Updated from global requirements","message":"Updated from global requirements\n\nChange-Id: I78c77c005a22f97613e13a6b2b3f7ab73ae01859\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/fc1c3ee3029b5175e329a6ec7639cec2f2d994c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/fc1c3ee3029b5175e329a6ec7639cec2f2d994c9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"674d304ee7865138be23f7183aede125e3526496","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
