)]}'
{"id":"openstack%2Finstack-undercloud~515123","triplet_id":"openstack%2Finstack-undercloud~master~I78e0671024b93ec64740cd7a8cda227a5145b789","project":"openstack/instack-undercloud","branch":"master","topic":"bug/1727406","hashtags":[],"change_id":"I78e0671024b93ec64740cd7a8cda227a5145b789","subject":"Set notification_bind on zaqar websocket","status":"MERGED","created":"2017-10-25 16:18:25.000000000","updated":"2017-10-26 08:27:29.000000000","submitted":"2017-10-26 06:20:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"515123-1508998830850-4e585a16","meta_rev_id":"e11361940f68db53a64074bdb5d029b4dbc7b386","_number":515123,"virtual_id_number":515123,"owner":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":2,"date":"2017-10-26 06:20:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2017-10-26 03:51:36.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2017-10-25 21:00:07.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2017-10-25 16:19:19.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2017-10-25 20:49:09.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2017-10-26 08:25:59.000000000","post_submit":true,"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"value":0,"date":"2017-10-26 04:19:13.000000000","_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-10-25 20:49:09.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2017-10-25 21:00:07.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2017-10-26 03:51:36.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-26 04:19:13.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2017-10-26 06:20:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2017-10-26 08:25:59.000000000","updated_by":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"reviewer":{"_account_id":360,"name":"Dan Prince","email":"dprince@redhat.com","username":"dan-prince"},"state":"REVIEWER"}],"messages":[{"id":"8f72ea87136e0d6e0fdeab2b0c7b06b645e6b028","author":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"date":"2017-10-25 16:18:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfe7f7e2cbafd1f0429e948b956a1469e1a90c8d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-10-25 16:19:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c492354df00e0d59546b7d5cf9d91ceedc79f0a7","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-10-25 18:10:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master-nv https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master-nv/503/ : SUCCESS in 1h 50m 12s (non-voting)\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024-master-nv https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024-master-nv/115/ : SUCCESS in 1h 52m 03s (non-voting)\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset022-master-nv https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset022-master-nv/117/ : SUCCESS in 1h 40m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a2cb464a69e2484404284c42ca1ac3374b2bc85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-25 19:14:59.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/23/515123/1/check-tripleo/legacy-tripleo-ci-centos-7-ovb-ha-oooq/c30868f/ : FAILURE in 2h 52m 54s\n- legacy-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/23/515123/1/check-tripleo/legacy-tripleo-ci-centos-7-ovb-containers-oooq/8d65c7f/ : SUCCESS in 2h 34m 22s\n- legacy-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/23/515123/1/check-tripleo/legacy-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/7eb7e72/ : SUCCESS in 2h 48m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e224553445ca3f0d9ab3bc50272710e5369f899a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-25 19:28:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/515123/1/check/openstack-tox-pep8/ef7473e/ : SUCCESS in 3m 08s\n- openstack-tox-py27 http://logs.openstack.org/23/515123/1/check/openstack-tox-py27/55d2e52/ : SUCCESS in 3m 23s\n- openstack-tox-py35 http://logs.openstack.org/23/515123/1/check/openstack-tox-py35/4cf55c7/ : SUCCESS in 3m 31s\n- legacy-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/23/515123/1/check/legacy-tripleo-ci-centos-7-undercloud-oooq/126a4ee/ : SUCCESS in 1h 02m 08s\n- legacy-tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/23/515123/1/check/legacy-tripleo-ci-centos-7-undercloud-upgrades/fb8dafc/ : SUCCESS in 58m 15s (non-voting)\n- legacy-puppet-syntax-3-centos-7 http://logs.openstack.org/23/515123/1/check/legacy-puppet-syntax-3-centos-7/7284959/ : SUCCESS in 8m 31s\n- legacy-puppet-syntax-4-centos-7 http://logs.openstack.org/23/515123/1/check/legacy-puppet-syntax-4-centos-7/9234b68/ : SUCCESS in 6m 39s\n- legacy-instack-undercloud-puppet-lint http://logs.openstack.org/23/515123/1/check/legacy-instack-undercloud-puppet-lint/776068d/ : SUCCESS in 6m 34s\n- openstack-tox-cover http://logs.openstack.org/23/515123/1/check/openstack-tox-cover/8878aaa/cover/ : SUCCESS in 3m 15s (non-voting)\n- legacy-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/23/515123/1/check/legacy-tripleo-ci-centos-7-nonha-multinode-oooq/6e9d130/ : SUCCESS in 1h 59m 32s\n- legacy-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/23/515123/1/check/legacy-tripleo-ci-centos-7-containers-multinode/b08f30f/ : SUCCESS in 1h 57m 30s\n- legacy-tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/23/515123/1/check/legacy-tripleo-ci-centos-7-undercloud-containers/bb2b27b/ : SUCCESS in 58m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f87aaf7e0162ca5e77c88557198871552327e5e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-10-25 20:49:09.000000000","message":"Patch Set 1: Workflow+1\n\nApproving this one to stabilize our CI. Please let me know if I did wrong. Trying to kill fires here.","accounts_in_message":[],"_revision_number":1},{"id":"daa1964db1e1fbb8ad037ddcae2a32e655b58149","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-25 20:49:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"077ba6a690f7979dff29cfea1f4853121ca80dd2","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2017-10-25 21:00:07.000000000","message":"Patch Set 1: Code-Review+2\n\nWould be nice to see all the ovb jobs pass, but it seems unlikely that this patch broke just one of them and ovb-ha has had timeout issues for a while.  Since this is breaking the gate I\u0027m good with merging it.","accounts_in_message":[],"_revision_number":1},{"id":"3b7ea4f17dab503f5b3ed980779fbd3797cefe41","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-10-25 21:06:12.000000000","message":"Patch Set 1:\n\nThanks Ben.","accounts_in_message":[],"_revision_number":1},{"id":"59630139eb10993478da1c7ccd76917aa24b16ec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-25 23:47:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/23/515123/1/gate/openstack-tox-pep8/44ed0e6/ : SUCCESS in 2m 47s\n- openstack-tox-py27 http://logs.openstack.org/23/515123/1/gate/openstack-tox-py27/1676650/ : SUCCESS in 3m 40s\n- openstack-tox-py35 http://logs.openstack.org/23/515123/1/gate/openstack-tox-py35/e830ee0/ : SUCCESS in 4m 16s\n- legacy-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/23/515123/1/gate/legacy-tripleo-ci-centos-7-undercloud-oooq/65e2ae8/ : SUCCESS in 1h 10m 40s\n- legacy-puppet-syntax-3-centos-7 http://logs.openstack.org/23/515123/1/gate/legacy-puppet-syntax-3-centos-7/3202a74/ : SUCCESS in 6m 03s\n- legacy-puppet-syntax-4-centos-7 http://logs.openstack.org/23/515123/1/gate/legacy-puppet-syntax-4-centos-7/6c30d8a/ : SUCCESS in 9m 33s\n- legacy-instack-undercloud-puppet-lint http://logs.openstack.org/23/515123/1/gate/legacy-instack-undercloud-puppet-lint/0f32452/ : SUCCESS in 3m 33s\n- legacy-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/23/515123/1/gate/legacy-tripleo-ci-centos-7-nonha-multinode-oooq/100efeb/ : FAILURE in 2h 56m 05s\n- legacy-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/23/515123/1/gate/legacy-tripleo-ci-centos-7-containers-multinode/0504840/ : SUCCESS in 1h 56m 48s","accounts_in_message":[],"_revision_number":1},{"id":"37c7bd4daa1bab0c7c3822fa22432b6aa575a2b6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2017-10-26 01:53:48.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e06cdca1bc331b43429354c33fbe89b22be4dc48","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2017-10-26 03:51:36.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master-nv https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master-nv/547/ : SUCCESS in 1h 57m 25s (non-voting)\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024-master-nv https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024-master-nv/159/ : SUCCESS in 1h 55m 12s (non-voting)\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset022-master-nv https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset022-master-nv/161/ : SUCCESS in 1h 49m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d3f0a7e160087396471ac4433b9432abbb4f10f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-26 03:56:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/515123/1/check/openstack-tox-pep8/66782f6/ : SUCCESS in 3m 47s\n- openstack-tox-py27 http://logs.openstack.org/23/515123/1/check/openstack-tox-py27/89887cd/ : SUCCESS in 3m 19s\n- openstack-tox-py35 http://logs.openstack.org/23/515123/1/check/openstack-tox-py35/28fc30a/ : SUCCESS in 4m 24s\n- legacy-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/23/515123/1/check/legacy-tripleo-ci-centos-7-undercloud-oooq/3831dfd/ : SUCCESS in 1h 01m 34s\n- legacy-tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/23/515123/1/check/legacy-tripleo-ci-centos-7-undercloud-upgrades/e911832/ : SUCCESS in 59m 10s (non-voting)\n- legacy-puppet-syntax-3-centos-7 http://logs.openstack.org/23/515123/1/check/legacy-puppet-syntax-3-centos-7/e14c065/ : SUCCESS in 6m 12s\n- legacy-puppet-syntax-4-centos-7 http://logs.openstack.org/23/515123/1/check/legacy-puppet-syntax-4-centos-7/e7ebc71/ : SUCCESS in 4m 54s\n- legacy-instack-undercloud-puppet-lint http://logs.openstack.org/23/515123/1/check/legacy-instack-undercloud-puppet-lint/fd6984e/ : SUCCESS in 3m 20s\n- openstack-tox-cover http://logs.openstack.org/23/515123/1/check/openstack-tox-cover/81bee4a/cover/ : SUCCESS in 3m 33s (non-voting)\n- legacy-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/23/515123/1/check/legacy-tripleo-ci-centos-7-nonha-multinode-oooq/22b2e1b/ : SUCCESS in 2h 00m 22s\n- legacy-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/23/515123/1/check/legacy-tripleo-ci-centos-7-containers-multinode/44d76ba/ : SUCCESS in 1h 46m 19s\n- legacy-tripleo-ci-centos-7-undercloud-containers http://logs.openstack.org/23/515123/1/check/legacy-tripleo-ci-centos-7-undercloud-containers/a91f6b7/ : SUCCESS in 59m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f2535dc403bdd864d8bcaee74be86fed067c691","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-26 03:56:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4f35ffe7ff92bfbd84f1399115f62e6a5bb6cad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-26 04:48:57.000000000","message":"Patch Set 1:\n\nBuild failed (check-tripleo pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- legacy-tripleo-ci-centos-7-ovb-ha-oooq http://logs.openstack.org/23/515123/1/check-tripleo/legacy-tripleo-ci-centos-7-ovb-ha-oooq/f09198f/ : FAILURE in 2h 53m 29s\n- legacy-tripleo-ci-centos-7-ovb-containers-oooq http://logs.openstack.org/23/515123/1/check-tripleo/legacy-tripleo-ci-centos-7-ovb-containers-oooq/722631c/ : SUCCESS in 2h 39m 38s\n- legacy-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024 http://logs.openstack.org/23/515123/1/check-tripleo/legacy-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024/a163da8/ : SUCCESS in 2h 51m 27s","accounts_in_message":[],"_revision_number":1},{"id":"09fbcb88e5539178bbf989398f058656bb667939","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-26 06:20:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/23/515123/1/gate/openstack-tox-pep8/9659ed5/ : SUCCESS in 2m 56s\n- openstack-tox-py27 http://logs.openstack.org/23/515123/1/gate/openstack-tox-py27/35e844e/ : SUCCESS in 4m 07s\n- openstack-tox-py35 http://logs.openstack.org/23/515123/1/gate/openstack-tox-py35/c3a7b0d/ : SUCCESS in 3m 12s\n- legacy-tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/23/515123/1/gate/legacy-tripleo-ci-centos-7-undercloud-oooq/5b243dc/ : SUCCESS in 1h 00m 09s\n- legacy-puppet-syntax-3-centos-7 http://logs.openstack.org/23/515123/1/gate/legacy-puppet-syntax-3-centos-7/9e9293b/ : SUCCESS in 6m 24s\n- legacy-puppet-syntax-4-centos-7 http://logs.openstack.org/23/515123/1/gate/legacy-puppet-syntax-4-centos-7/4f0a5bf/ : SUCCESS in 6m 07s\n- legacy-instack-undercloud-puppet-lint http://logs.openstack.org/23/515123/1/gate/legacy-instack-undercloud-puppet-lint/9909b1b/ : SUCCESS in 3m 33s\n- legacy-tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/23/515123/1/gate/legacy-tripleo-ci-centos-7-nonha-multinode-oooq/0bea2d0/ : SUCCESS in 2h 19m 08s\n- legacy-tripleo-ci-centos-7-containers-multinode http://logs.openstack.org/23/515123/1/gate/legacy-tripleo-ci-centos-7-containers-multinode/383f5e7/ : SUCCESS in 1h 57m 20s","accounts_in_message":[],"_revision_number":1},{"id":"3fdee52867ac89a6140a0319ea68976d4d5529f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-10-26 06:20:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7323d0324bd6cd5ccc5c07ccdf88684b69f8ce7e","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2017-10-26 06:37:22.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/pike as commit 211eb5bc6c4a019e0b70c99697cc66614f68c1d1","accounts_in_message":[],"_revision_number":1},{"id":"b4cffb88dcec5896ea95f9a68d4df389d418b800","author":{"_account_id":6926,"name":"Bogdan Dobrelya","email":"bdobreli@redhat.com","username":"bogdando"},"date":"2017-10-26 08:26:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"218b7ed56d0dfd5d4607052782f051617685cf60","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2017-10-26 08:27:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5fdb4d7913bfc0096c725f6f99fd98e949c75d58","revisions":{"5fdb4d7913bfc0096c725f6f99fd98e949c75d58":{"kind":"REWORK","_number":1,"created":"2017-10-25 16:18:25.000000000","uploader":{"_account_id":7385,"name":"Thomas Herve","email":"therve@redhat.com","username":"therve"},"ref":"refs/changes/23/515123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/23/515123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/23/515123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/23/515123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/23/515123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/23/515123/1"}}},"commit":{"parents":[{"commit":"031fa7ea3ecf11e8f7cfdb32b60106588f1b4627","subject":"Merge \"Remove Pike-only migration step\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/031fa7ea3ecf11e8f7cfdb32b60106588f1b4627"}]}],"author":{"name":"Thomas Herve","email":"therve@redhat.com","date":"2017-10-25 16:17:06.000000000","tz":120},"committer":{"name":"Thomas Herve","email":"therve@redhat.com","date":"2017-10-25 16:17:06.000000000","tz":120},"subject":"Set notification_bind on zaqar websocket","message":"Set notification_bind on zaqar websocket\n\nDon\u0027t let zaqar websocket listen to notifications on all interfaces:\notherwise, it uses the hostname as a subscription target, and it doesn\u0027t\nwork when the host is set as an IPv6 host. Ultimately this should be\nfixed in Zaqar, but this will work around it for now.\n\nChange-Id: I78e0671024b93ec64740cd7a8cda227a5145b789\nRelated-Bug: #1727406\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/5fdb4d7913bfc0096c725f6f99fd98e949c75d58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/5fdb4d7913bfc0096c725f6f99fd98e949c75d58"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"031fa7ea3ecf11e8f7cfdb32b60106588f1b4627","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
