)]}'
{"id":"openstack%2Finstack-undercloud~536570","triplet_id":"openstack%2Finstack-undercloud~stable%2Focata~I89828f9852e8d35878aa8eb0a16ae0ece9418575","project":"openstack/instack-undercloud","branch":"stable/ocata","topic":"bug/1744763-stable/ocata","hashtags":[],"change_id":"I89828f9852e8d35878aa8eb0a16ae0ece9418575","subject":"Remove RH1 OVB jobs from configuration","status":"MERGED","created":"2018-01-22 20:42:10.000000000","updated":"2018-01-24 11:45:39.000000000","submitted":"2018-01-24 11:45:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"536570-1516794339912-217fbfef","meta_rev_id":"f31abd19a1faa707d8d783600fbef2532701b8f9","_number":536570,"virtual_id_number":536570,"owner":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":2,"date":"2018-01-24 11:45:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2018-01-23 01:51:14.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2018-01-23 00:53:03.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2018-01-23 15:35:56.000000000","_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-23 00:53:03.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2018-01-23 01:51:14.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-01-24 11:45:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6d0d5917b5f18c59b050176359193b79f4886fcb","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2018-01-22 20:42:10.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c5faeef31a2908c70807edc28f01bd5f455cc28b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-22 23:16:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/70/536570/1/check/openstack-tox-pep8/cd61745/ : SUCCESS in 2m 45s\n- openstack-tox-py27 http://logs.openstack.org/70/536570/1/check/openstack-tox-py27/622dc3b/ : SUCCESS in 3m 05s\n- openstack-tox-py35 http://logs.openstack.org/70/536570/1/check/openstack-tox-py35/f5fd106/ : SUCCESS in 2m 54s\n- openstack-tox-cover http://logs.openstack.org/70/536570/1/check/openstack-tox-cover/e8e9ebd/cover/ : SUCCESS in 3m 02s (non-voting)\n- puppet-openstack-lint http://logs.openstack.org/70/536570/1/check/puppet-openstack-lint/42c294d/ : SUCCESS in 4m 01s\n- puppet-openstack-syntax-4 http://logs.openstack.org/70/536570/1/check/puppet-openstack-syntax-4/9b46a34/ : SUCCESS in 4m 05s\n- puppet-openstack-syntax-3-legacy http://logs.openstack.org/70/536570/1/check/puppet-openstack-syntax-3-legacy/04bfb0e/ : SUCCESS in 3m 34s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/70/536570/1/check/tripleo-ci-centos-7-undercloud-oooq/55b97e0/ : SUCCESS in 1h 08m 18s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/70/536570/1/check/tripleo-ci-centos-7-undercloud-upgrades/bab5577/ : SUCCESS in 1h 54m 36s (non-voting)\n- tripleo-ci-centos-7-3nodes-multinode http://logs.openstack.org/70/536570/1/check/tripleo-ci-centos-7-3nodes-multinode/738ed87/ : FAILURE in 1h 36m 17s (non-voting)\n- tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/70/536570/1/check/tripleo-ci-centos-7-nonha-multinode-oooq/c239fa4/ : SUCCESS in 2h 00m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d104112501161dc41b0f528f5ef0f217c84bdb08","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-01-23 00:53:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6e1ef929de35d706c49cbc31eccb235d479564e6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-01-23 01:51:14.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-ocata/61/ : SUCCESS in 1h 45m 21s\n- gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024-ocata https://review.rdoproject.org/jenkins/job/gate-tripleo-ci-centos-7-ovb-1ctlr_1comp_1ceph-featureset024-ocata/62/ : FAILURE in 1h 46m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f48dfdaef5d8196d98a6f31da5f911bb8a55c4e7","author":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"date":"2018-01-23 15:35:56.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71e239ccc4f038f1fa251c48765d4119d6fd9642","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-24 05:18:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"be3f4b315cee03ca56afd4e503506a96529ca5b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-24 11:45:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/70/536570/1/gate/openstack-tox-pep8/8b4fab9/ : SUCCESS in 3m 07s\n- openstack-tox-py27 http://logs.openstack.org/70/536570/1/gate/openstack-tox-py27/1a92883/ : SUCCESS in 3m 34s\n- openstack-tox-py35 http://logs.openstack.org/70/536570/1/gate/openstack-tox-py35/c23ed06/ : SUCCESS in 3m 35s\n- puppet-openstack-lint http://logs.openstack.org/70/536570/1/gate/puppet-openstack-lint/9ffdb05/ : SUCCESS in 5m 16s\n- puppet-openstack-syntax-4 http://logs.openstack.org/70/536570/1/gate/puppet-openstack-syntax-4/e46151d/ : SUCCESS in 3m 55s\n- puppet-openstack-syntax-3-legacy http://logs.openstack.org/70/536570/1/gate/puppet-openstack-syntax-3-legacy/8ddae4c/ : SUCCESS in 3m 43s\n- tripleo-ci-centos-7-undercloud-oooq http://logs.openstack.org/70/536570/1/gate/tripleo-ci-centos-7-undercloud-oooq/6aca9c1/ : SUCCESS in 1h 09m 16s\n- tripleo-ci-centos-7-nonha-multinode-oooq http://logs.openstack.org/70/536570/1/gate/tripleo-ci-centos-7-nonha-multinode-oooq/02b746a/ : SUCCESS in 1h 57m 08s","accounts_in_message":[],"_revision_number":1},{"id":"a2c5cff88fc3e7d118faa3648b09536bf0aeacd7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-24 11:45:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ded9f9e9bdc5b4c4e324acc21c3d391a59111c13","revisions":{"ded9f9e9bdc5b4c4e324acc21c3d391a59111c13":{"kind":"REWORK","_number":1,"created":"2018-01-22 20:42:10.000000000","uploader":{"_account_id":10969,"name":"Shnaidman Sagi (Sergey)","display_name":"Shnaidman Sagi","email":"sshnaidm@redhat.com","username":"sergsh"},"ref":"refs/changes/70/536570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/70/536570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/70/536570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/70/536570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/70/536570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/70/536570/1"}}},"commit":{"parents":[{"commit":"7e65c96eceb2794c4d4f7fb5fbafe6188408f3a4","subject":"Set password for mysql root user on undercloud","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/7e65c96eceb2794c4d4f7fb5fbafe6188408f3a4"}]}],"author":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2018-01-22 17:32:46.000000000","tz":120},"committer":{"name":"Sagi Shnaidman","email":"sshnaidm@redhat.com","date":"2018-01-22 20:42:10.000000000","tz":0},"subject":"Remove RH1 OVB jobs from configuration","message":"Remove RH1 OVB jobs from configuration\n\nRemove OVB RH1 jobs from configuration for this repo\nRelated-Bug: #1744763\n\nChange-Id: I89828f9852e8d35878aa8eb0a16ae0ece9418575\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/ded9f9e9bdc5b4c4e324acc21c3d391a59111c13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/ded9f9e9bdc5b4c4e324acc21c3d391a59111c13"}]},"parents_data":[{"branch_name":"refs/heads/stable/ocata","commit_id":"7e65c96eceb2794c4d4f7fb5fbafe6188408f3a4","is_merged_in_target_branch":true,"change_id":"I408ce3a0fe2ab8e86bcc280256cdb51688efde75","change_number":532806,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/stable/ocata"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
