)]}'
{"id":"openstack%2Finstack-undercloud~605755","triplet_id":"openstack%2Finstack-undercloud~master~Ia5bc6dcea16bde78bb47747b8c0ed736f4f576bb","project":"openstack/instack-undercloud","branch":"master","topic":"remove-instack-undercloud","hashtags":[],"change_id":"Ia5bc6dcea16bde78bb47747b8c0ed736f4f576bb","subject":"Add instack-undercloud job","status":"ABANDONED","created":"2018-09-27 14:35:07.000000000","updated":"2018-10-26 19:04:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3b53b9e632b66531f55cee4892dd117fce9a87ab","_number":605755,"virtual_id_number":605755,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":-1,"date":"2018-10-18 14:52:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2018-10-18 13:51:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2018-10-24 19:07:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"date":"2018-10-24 19:07:52.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":-1,"date":"2018-10-25 14:30:44.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-09-27 21:49:58.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2018-09-28 02:03:35.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-10-18 14:52:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2018-10-24 19:07:52.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"}],"messages":[{"id":"f0b8ec646f29722efbbc85c3ba3e5456b4e4d419","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-09-27 14:35:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ce487ce4efbd96d86a5f8bdc80d0181d34062923","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-27 20:39:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/55/605755/1/check/openstack-tox-pep8/11864b0/ : SUCCESS in 3m 16s\n- openstack-tox-py27 http://logs.openstack.org/55/605755/1/check/openstack-tox-py27/cd57b81/ : SUCCESS in 3m 31s\n- openstack-tox-py35 http://logs.openstack.org/55/605755/1/check/openstack-tox-py35/a8dd68a/ : SUCCESS in 3m 43s\n- openstack-tox-cover http://logs.openstack.org/55/605755/1/check/openstack-tox-cover/f2b6e00/cover/ : SUCCESS in 3m 31s (non-voting)\n- puppet-openstack-lint http://logs.openstack.org/55/605755/1/check/puppet-openstack-lint/a7a1f95/ : SUCCESS in 3m 59s\n- puppet-openstack-syntax-4 http://logs.openstack.org/55/605755/1/check/puppet-openstack-syntax-4/5d83de7/ : SUCCESS in 3m 33s\n- puppet-openstack-syntax-5 http://logs.openstack.org/55/605755/1/check/puppet-openstack-syntax-5/250fa0a/ : SUCCESS in 5m 55s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/55/605755/1/check/tripleo-ci-centos-7-undercloud-upgrades/949336e/ : SUCCESS in 1h 23m 25s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/605755/1/check/openstack-tox-lower-constraints/dc10801/ : SUCCESS in 4m 32s\n- openstack-tox-py36 http://logs.openstack.org/55/605755/1/check/openstack-tox-py36/511eca6/ : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"64a5951028624669d69211156a5e6d11d2cc6ae2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-09-27 21:34:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0824e45cb3f738e48d39f06e318dec0d6a32b179","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-09-27 21:37:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8c5beb4ac898abf335a6c509f6af9529ca01dbe7","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-09-27 21:49:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d3ca04029d8d33999a7b5f796a3a4b6ebcda8faa","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2018-09-28 02:03:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master http://logs.rdoproject.org/55/605755/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001-master/2b1601c/ : SUCCESS in 2h 43m 47s\n- legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master http://logs.rdoproject.org/55/605755/3/openstack-check/legacy-tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035-master/d559ab6/ : SUCCESS in 2h 35m 13s","accounts_in_message":[],"_revision_number":3},{"id":"9d3c051c9edb204719e2dc795b1c7e9cdb0ae19d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-09-28 03:23:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/55/605755/3/check/openstack-tox-pep8/0ff6f6e/ : SUCCESS in 2m 55s\n- openstack-tox-py27 http://logs.openstack.org/55/605755/3/check/openstack-tox-py27/fcd2fd2/ : SUCCESS in 3m 28s\n- openstack-tox-py35 http://logs.openstack.org/55/605755/3/check/openstack-tox-py35/be0f6fa/ : SUCCESS in 3m 14s\n- openstack-tox-cover http://logs.openstack.org/55/605755/3/check/openstack-tox-cover/d623b33/cover/ : SUCCESS in 3m 30s (non-voting)\n- puppet-openstack-lint http://logs.openstack.org/55/605755/3/check/puppet-openstack-lint/37af71b/ : SUCCESS in 3m 47s\n- puppet-openstack-syntax-4 http://logs.openstack.org/55/605755/3/check/puppet-openstack-syntax-4/a5f1238/ : SUCCESS in 4m 03s\n- puppet-openstack-syntax-5 http://logs.openstack.org/55/605755/3/check/puppet-openstack-syntax-5/5aa35ec/ : SUCCESS in 6m 05s\n- tripleo-ci-centos-7-instack-undercloud-oooq http://logs.openstack.org/55/605755/3/check/tripleo-ci-centos-7-instack-undercloud-oooq/3f9b2af/ : FAILURE in 34m 03s\n- tripleo-ci-centos-7-undercloud-upgrades http://logs.openstack.org/55/605755/3/check/tripleo-ci-centos-7-undercloud-upgrades/4cb0445/ : SUCCESS in 1h 10m 23s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/605755/3/check/openstack-tox-lower-constraints/fa979dd/ : SUCCESS in 4m 34s\n- openstack-tox-py36 http://logs.openstack.org/55/605755/3/check/openstack-tox-py36/a8a12c0/ : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b241be87158cd49e88453f51ade3a60f85e99fe6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2018-10-18 13:51:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"a1106acff8f8eaff7eb6fc0deab8daf6d1e37d02","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-18 14:52:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/55/605755/4/check/puppet-openstack-lint/f961a93/ : SUCCESS in 4m 12s\n- puppet-openstack-syntax-4 http://logs.openstack.org/55/605755/4/check/puppet-openstack-syntax-4/f9be1ca/ : SUCCESS in 3m 36s\n- puppet-openstack-syntax-5 http://logs.openstack.org/55/605755/4/check/puppet-openstack-syntax-5/c3706b4/ : SUCCESS in 4m 41s\n- tripleo-ci-centos-7-instack-undercloud-oooq http://logs.openstack.org/55/605755/4/check/tripleo-ci-centos-7-instack-undercloud-oooq/e607db0/ : FAILURE in 20m 23s\n- openstack-tox-cover http://logs.openstack.org/55/605755/4/check/openstack-tox-cover/692eb69/cover/ : SUCCESS in 2m 59s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/605755/4/check/openstack-tox-lower-constraints/67d7bfa/ : SUCCESS in 5m 21s\n- openstack-tox-pep8 http://logs.openstack.org/55/605755/4/check/openstack-tox-pep8/fb16fe2/ : SUCCESS in 3m 11s\n- openstack-tox-py27 http://logs.openstack.org/55/605755/4/check/openstack-tox-py27/faa9558/ : SUCCESS in 3m 32s\n- openstack-tox-py35 http://logs.openstack.org/55/605755/4/check/openstack-tox-py35/e5dd4c9/ : SUCCESS in 3m 56s\n- openstack-tox-py36 http://logs.openstack.org/55/605755/4/check/openstack-tox-py36/c5fbada/ : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":4},{"id":"4964efd9dc71195bc1485927b98977a7b6fe7c12","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2018-10-24 19:07:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f830769afdbe1a49fb5cdfc4beeb3919a5f52482","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-10-24 22:14:13.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint http://logs.openstack.org/55/605755/4/check/puppet-openstack-lint/34d1ce8/ : SUCCESS in 3m 54s\n- puppet-openstack-syntax-4 http://logs.openstack.org/55/605755/4/check/puppet-openstack-syntax-4/9e164c4/ : SUCCESS in 4m 10s\n- puppet-openstack-syntax-5 http://logs.openstack.org/55/605755/4/check/puppet-openstack-syntax-5/678c1e9/ : SUCCESS in 4m 19s\n- tripleo-ci-centos-7-instack-undercloud-oooq http://logs.openstack.org/55/605755/4/check/tripleo-ci-centos-7-instack-undercloud-oooq/2fd0565/ : FAILURE in 19m 16s\n- openstack-tox-cover http://logs.openstack.org/55/605755/4/check/openstack-tox-cover/78ef15b/cover/ : SUCCESS in 3m 29s\n- openstack-tox-lower-constraints http://logs.openstack.org/55/605755/4/check/openstack-tox-lower-constraints/41a24ba/ : FAILURE in 4m 27s\n- openstack-tox-pep8 http://logs.openstack.org/55/605755/4/check/openstack-tox-pep8/39dce57/ : FAILURE in 2m 57s\n- openstack-tox-py27 http://logs.openstack.org/55/605755/4/check/openstack-tox-py27/87e394d/ : SUCCESS in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/55/605755/4/check/openstack-tox-py35/be8bb9c/ : SUCCESS in 3m 35s\n- openstack-tox-py36 http://logs.openstack.org/55/605755/4/check/openstack-tox-py36/0907248/ : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":4},{"id":"b288f4ec95530632288f0eb106c617d540a5a57f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-10-25 14:30:44.000000000","message":"Patch Set 4: Workflow-1\n\nI need to fix this i think, something is missing","accounts_in_message":[],"_revision_number":4},{"id":"d5b0b7c18ac37d68cd52c83691c85664d39b9129","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-10-25 14:30:52.000000000","message":"Removed Workflow+1 by James Slagle \u003cjslagle@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":4},{"id":"f31abd3078014fe347489fd31aaf6cc5be29a2b8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2018-10-26 19:04:15.000000000","message":"Abandoned\n\nLet\u0027s abandon this as instack_undercloud is not used in master anymore.  Patches should go directly in stable/* branches.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f5d5b7123501315895b85694ed1771cffce7d07a","revisions":{"2605c67c2836fefdb867bb9df4861ef62b042993":{"kind":"REWORK","_number":1,"created":"2018-09-27 14:35:07.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/605755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/55/605755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/1"}}},"commit":{"parents":[{"commit":"c220b84ef799960b3e4801940f5b5c77522840fa","subject":"Merge \"Use templates for cover and lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/c220b84ef799960b3e4801940f5b5c77522840fa"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-09-27 14:34:53.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-09-27 14:34:53.000000000","tz":-360},"subject":"DNM: testing ci","message":"DNM: testing ci\n\nChange-Id: Ia5bc6dcea16bde78bb47747b8c0ed736f4f576bb\nDepends-On: https://review.openstack.org/#/c/605753/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/2605c67c2836fefdb867bb9df4861ef62b042993"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/2605c67c2836fefdb867bb9df4861ef62b042993"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c220b84ef799960b3e4801940f5b5c77522840fa","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"a73119937d6f81664ddb8b0d90e6462bf4d7ab4b":{"kind":"REWORK","_number":2,"created":"2018-09-27 21:34:27.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/605755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/55/605755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/2"}}},"commit":{"parents":[{"commit":"c220b84ef799960b3e4801940f5b5c77522840fa","subject":"Merge \"Use templates for cover and lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/c220b84ef799960b3e4801940f5b5c77522840fa"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-09-27 14:34:53.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-09-27 21:33:53.000000000","tz":-360},"subject":"Add instack-undercloud job","message":"Add instack-undercloud job\n\nWe removed the instack-undercloud job for anything newer than Rocky. We\nstill need to run this job on the instack-undercloud project so let\u0027s\nadd it in here.\n\nChange-Id: Ia5bc6dcea16bde78bb47747b8c0ed736f4f576bb\nDepends-On: https://review.openstack.org/#/c/605753/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/a73119937d6f81664ddb8b0d90e6462bf4d7ab4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/a73119937d6f81664ddb8b0d90e6462bf4d7ab4b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c220b84ef799960b3e4801940f5b5c77522840fa","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"318fbe4272b22b52e87e55207631d368a444fcf9":{"kind":"REWORK","_number":3,"created":"2018-09-27 21:37:37.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/55/605755/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/55/605755/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/3"}}},"commit":{"parents":[{"commit":"c220b84ef799960b3e4801940f5b5c77522840fa","subject":"Merge \"Use templates for cover and lower-constraints\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/c220b84ef799960b3e4801940f5b5c77522840fa"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-09-27 14:34:53.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-09-27 21:37:34.000000000","tz":-360},"subject":"Add instack-undercloud job","message":"Add instack-undercloud job\n\nWe removed the instack-undercloud job for anything newer than Rocky. We\nstill need to run this job on the instack-undercloud project so let\u0027s\nadd it in here.\n\nChange-Id: Ia5bc6dcea16bde78bb47747b8c0ed736f4f576bb\nDepends-On: https://review.openstack.org/#/c/605753/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/318fbe4272b22b52e87e55207631d368a444fcf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/318fbe4272b22b52e87e55207631d368a444fcf9"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c220b84ef799960b3e4801940f5b5c77522840fa","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"f5d5b7123501315895b85694ed1771cffce7d07a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2018-10-18 13:51:48.000000000","uploader":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"ref":"refs/changes/55/605755/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/55/605755/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/55/605755/4"}}},"commit":{"parents":[{"commit":"8c0a8316ce62f4e64d5088e1c78853a54612bf48","subject":"Merge \"Update the outdated URL in doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/8c0a8316ce62f4e64d5088e1c78853a54612bf48"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2018-09-27 14:34:53.000000000","tz":-360},"committer":{"name":"Emilien Macchi","email":"emilien@redhat.com","date":"2018-10-18 13:51:48.000000000","tz":0},"subject":"Add instack-undercloud job","message":"Add instack-undercloud job\n\nWe removed the instack-undercloud job for anything newer than Rocky. We\nstill need to run this job on the instack-undercloud project so let\u0027s\nadd it in here.\n\nChange-Id: Ia5bc6dcea16bde78bb47747b8c0ed736f4f576bb\nDepends-On: https://review.openstack.org/#/c/605753/\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/f5d5b7123501315895b85694ed1771cffce7d07a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/f5d5b7123501315895b85694ed1771cffce7d07a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"8c0a8316ce62f4e64d5088e1c78853a54612bf48","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
