)]}'
{"id":"openstack%2Finstack-undercloud~678087","triplet_id":"openstack%2Finstack-undercloud~stable%2Frocky~I75bb75092fa348dc1c2dc465446cb91d4882cf15","project":"openstack/instack-undercloud","branch":"stable/rocky","topic":"bug/1841100","hashtags":[],"change_id":"I75bb75092fa348dc1c2dc465446cb91d4882cf15","subject":"Explicitly set notification driver for novajoin","status":"MERGED","created":"2019-08-22 18:51:06.000000000","updated":"2019-08-23 14:03:52.000000000","submitted":"2019-08-23 05:19:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"678087-1566537568368-02e6e954","meta_rev_id":"4f2fe89d4acfee223d9171954dfd07e25c5748de","_number":678087,"virtual_id_number":678087,"owner":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":2,"date":"2019-08-23 05:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-22 21:52:54.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-08-22 19:53:47.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-22 19:29:04.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-08-22 19:29:19.000000000","_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-22 19:29:19.000000000","updated_by":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"reviewer":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"state":"REVIEWER"},{"updated":"2019-08-22 19:53:47.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-22 21:52:54.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-23 05:19:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a464d5fe5c2e9a165b32516189a91f939a44e27e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 18:51:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2e79e25c85c86abc94677b62f19e7beb3610963","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 19:16:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a6b775480f4a1200204cd9cfebaa915b231f08bd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 19:25:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2269da892e84942006614f9435284b9078ffb2b7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-22 19:27:39.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"3e0dba5f05f8630ebdbd09b0fc09e10b9c739275","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-22 19:29:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a368f217454717ed486a1e3536890a9a179a14c8","author":{"_account_id":7144,"name":"James Slagle","email":"jslagle@redhat.com","username":"slagle"},"date":"2019-08-22 19:29:19.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4d1a8008ad9af71f89a805721b1df70539532934","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-22 19:53:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"223327f3b61806dd09af2a86e758c5a73cf84cd4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 21:44:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/f399736dfa5d4093a3f90201a59f3776 : SUCCESS in 5m 19s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/ab05f60e96e5457aa7bad2ce15ae489b : SUCCESS in 5m 31s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/41c2d2b113404f0fab808b23abaf2b95 : SUCCESS in 5m 22s\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/d0210da37d55495c9df88db2d5876a6e : SUCCESS in 1h 36m 45s\n- tripleo-ci-centos-7-undercloud-upgrades https://zuul.opendev.org/t/openstack/build/ff5d42007acc47f885aaa44d6d8cfe9a : SUCCESS in 2h 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/892ce8274bc84ed58442540799aa78e2 : SUCCESS in 5m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72afc7d8b87e432288d95a7cc7ab91ca : SUCCESS in 3m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d110d27a12ce45aabd0439b63133a448 : SUCCESS in 3m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7a449da4cb24ad6b26366851de792b9 : SUCCESS in 5m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19ca1607667e44fea3614bac3ea1f612 : SUCCESS in 3m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6d8e646ce2b120f6052388b6dd9234c809e4fae2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-22 21:45:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8ec9fb51fafe181a079a0dad08242c9d4a33bcec","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-22 21:52:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/87/678087/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/509efb3/ : SUCCESS in 2h 24m 37s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/87/678087/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/83b6c02/ : SUCCESS in 2h 13m 45s","accounts_in_message":[],"_revision_number":3},{"id":"f97ce3ad68b57532de25fc87c992a42db961888f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 05:19:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f5ac9ba64827425000a73c7e2842257310faccbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-23 05:19:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint https://zuul.opendev.org/t/openstack/build/0d25842adf1f48848ad1cfe38f091d9b : SUCCESS in 5m 13s\n- puppet-openstack-syntax-4 https://zuul.opendev.org/t/openstack/build/eda6600ad02741aabdd5ac21708971ed : SUCCESS in 6m 57s\n- puppet-openstack-syntax-5 https://zuul.opendev.org/t/openstack/build/a4ce37c9e88843a898eba9f2d8305975 : SUCCESS in 5m 07s\n- tripleo-ci-centos-7-undercloud-oooq https://zuul.opendev.org/t/openstack/build/46270b5b01344e0b83e7a59f00c599a1 : SUCCESS in 1h 01m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b0747df6edc426bb07ee121400e1aea : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97492aa8874943d48b189826bd789330 : SUCCESS in 4m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b58bff7a2d5e433e9c9421b512a7875f : SUCCESS in 4m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/216ad088c72848d9982d5cb688d97618 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":3},{"id":"3a3af8ef2770aac6ba1656d1b70a2aed6afbf6c2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-23 14:03:52.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit 97b77d1d13321833c82fded6b817105e2624ab78","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5f1b6c6e5001ed917d474bed33ca88103031de94","revisions":{"c3e0c9bfcb0c6fbe08ec14e3c86803ae06fba662":{"kind":"REWORK","_number":1,"created":"2019-08-22 18:51:06.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/87/678087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/87/678087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/87/678087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/87/678087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/87/678087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/87/678087/1"}}},"commit":{"parents":[{"commit":"8b6d68d13fab7cf28388d7e48f2b89be2c8932e8","subject":"Set notification driver to noop when enable_telemetry\u003dfalse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/8b6d68d13fab7cf28388d7e48f2b89be2c8932e8"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-22 18:50:05.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-22 18:50:05.000000000","tz":-360},"subject":"Explicitly set notification driver for novajoin","message":"Explicitly set notification driver for novajoin\n\nNovajoin is reliant on some notifications from nova and neutron. We\nrecently merged an update to the undercloud installation that disables\nthe notification drvier across the board if telemetry is disabled. This\nchange adds logic for nova and neutron to force the notification driver\nto messagingv2.\n\nChange-Id: I75bb75092fa348dc1c2dc465446cb91d4882cf15\nCloses-Bug: #1841100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/c3e0c9bfcb0c6fbe08ec14e3c86803ae06fba662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/c3e0c9bfcb0c6fbe08ec14e3c86803ae06fba662"}]},"branch":"refs/heads/stable/rocky"},"4366699b21596848ff68ef62e105b0e0f635747b":{"kind":"REWORK","_number":2,"created":"2019-08-22 19:25:17.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/87/678087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/87/678087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/87/678087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/87/678087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/87/678087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/87/678087/2"}}},"commit":{"parents":[{"commit":"8b6d68d13fab7cf28388d7e48f2b89be2c8932e8","subject":"Set notification driver to noop when enable_telemetry\u003dfalse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/8b6d68d13fab7cf28388d7e48f2b89be2c8932e8"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-22 18:50:05.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-22 19:25:14.000000000","tz":-360},"subject":"Explicitly set notification driver for novajoin","message":"Explicitly set notification driver for novajoin\n\nNovajoin is reliant on some notifications from nova and neutron. We\nrecently merged an update to the undercloud installation that disables\nthe notification drvier across the board if telemetry is disabled. This\nchange adds logic for nova and neutron to force the notification driver\nto messagingv2.\n\nChange-Id: I75bb75092fa348dc1c2dc465446cb91d4882cf15\nCloses-Bug: #1841100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/4366699b21596848ff68ef62e105b0e0f635747b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/4366699b21596848ff68ef62e105b0e0f635747b"}]},"branch":"refs/heads/stable/rocky"},"5f1b6c6e5001ed917d474bed33ca88103031de94":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-08-22 19:27:39.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/87/678087/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/instack-undercloud","ref":"refs/changes/87/678087/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/87/678087/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/87/678087/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/instack-undercloud refs/changes/87/678087/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/instack-undercloud refs/changes/87/678087/3"}}},"commit":{"parents":[{"commit":"8b6d68d13fab7cf28388d7e48f2b89be2c8932e8","subject":"Set notification driver to noop when enable_telemetry\u003dfalse","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/8b6d68d13fab7cf28388d7e48f2b89be2c8932e8"}]}],"author":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-22 18:50:05.000000000","tz":-360},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-22 19:27:12.000000000","tz":-360},"subject":"Explicitly set notification driver for novajoin","message":"Explicitly set notification driver for novajoin\n\nNovajoin is reliant on some notifications from nova. We recently\nmerged an update to the undercloud installation that disables\nthe notification drvier across the board if telemetry is disabled. This\nchange adds logic for nova to force the notification driver to\nmessagingv2.\n\nChange-Id: I75bb75092fa348dc1c2dc465446cb91d4882cf15\nCloses-Bug: #1841100\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/5f1b6c6e5001ed917d474bed33ca88103031de94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/instack-undercloud/commit/5f1b6c6e5001ed917d474bed33ca88103031de94"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
