)]}'
{"id":"openstack%2Fironic-inspector~684367","triplet_id":"openstack%2Fironic-inspector~master~I18dbc1ab9296110d3a0a2d4400f81a3d0e74564c","project":"openstack/ironic-inspector","branch":"master","topic":"test-ci","hashtags":[],"change_id":"I18dbc1ab9296110d3a0a2d4400f81a3d0e74564c","subject":"[DNM] testing ironic-inspector CI","status":"ABANDONED","created":"2019-09-24 14:50:39.000000000","updated":"2019-10-04 08:44:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e539b4df22e44fb4fc8ab24f66cf348cc8630d96","_number":684367,"virtual_id_number":684367,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-09-24 17:44:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2019-09-24 17:49:19.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-24 17:44:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-09-24 17:49:19.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1017cb531e52d7399b197f452037bf2bacbe9cf2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-09-24 14:50:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52533ff45466da4678af4251655cece64737e246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-24 17:44:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e03c40055a474265b0060dd040387f37 : SUCCESS in 5m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/61416b49ca8845389dcb98622cbba34a : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f26cc7bd57264b7087ca009922f4ebb3 : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eeccba5808c5430eb2a2b9aa859efe96 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47061479bff442f5a4ef7a4935780281 : SUCCESS in 3m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5117dd12269448cd9e4b150b5e472c50 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6456eddc9d0b4db2aef163b54a2c506e : SUCCESS in 8m 28s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/d322d2f71f6942eb94b52cd6dbeb09b5 : SUCCESS in 2h 45m 36s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/08f95b593cc34c7f816bcfbe5fef8fcb : SUCCESS in 49m 15s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/09a835e83b274906a83c63d5ad53eff3 : SUCCESS in 43m 25s\n- ironic-inspector-tempest-python3 https://zuul.opendev.org/t/openstack/build/458a6837d399465ea598da2cb93935f0 : SUCCESS in 54m 25s\n- ironic-inspector-non-standalone-tempest-python3 https://zuul.opendev.org/t/openstack/build/1a12fc38ffcf4d50a204ee55d2d4e558 : SUCCESS in 55m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/00a7a96ca4c1408081045bf5c2587826 : FAILURE in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86cd922233d14e8b865b8e601ce3eb51 : FAILURE in 4m 20s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/61f7f2b7136d42b5a526d3bf05c71ece : FAILURE in 12m 03s\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/e7c628853a0b4d09b2bb162dd5563bab : SUCCESS in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ece6a04132179f3172b371dbdc8be0ef55727511","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-09-24 17:49:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/67/684367/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/c3e2029/ : SUCCESS in 2h 52m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b402ad279e045bf079fdc346dea878a2b000428c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-04 08:44:45.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f08d954b1183cd0019185cb51eddf8a361d41d30","revisions":{"f08d954b1183cd0019185cb51eddf8a361d41d30":{"kind":"REWORK","_number":1,"created":"2019-09-24 14:50:39.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/67/684367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/67/684367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/67/684367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/67/684367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/67/684367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/67/684367/1"}}},"commit":{"parents":[{"commit":"40f1dff1e57d2f90a8279c7631e2a502d9631b7f","subject":"Clean up release notes for 10.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/40f1dff1e57d2f90a8279c7631e2a502d9631b7f"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-09-24 14:49:01.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2019-09-24 14:50:37.000000000","tz":120},"subject":"[DNM] testing ironic-inspector CI","message":"[DNM] testing ironic-inspector CI\n\nChange-Id: I18dbc1ab9296110d3a0a2d4400f81a3d0e74564c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/f08d954b1183cd0019185cb51eddf8a361d41d30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/f08d954b1183cd0019185cb51eddf8a361d41d30"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
