)]}'
{"id":"openstack%2Fironic-inspector~697286","triplet_id":"openstack%2Fironic-inspector~stable%2Ftrain~I1378a1a4d2a35ba8f182138877b5151f25d4c56d","project":"openstack/ironic-inspector","branch":"stable/train","topic":"story_2006950-stable/train","hashtags":[],"change_id":"I1378a1a4d2a35ba8f182138877b5151f25d4c56d","subject":"Get details when listing ports upon inspection of active nodes","status":"MERGED","created":"2019-12-04 13:42:14.000000000","updated":"2019-12-05 02:10:38.000000000","submitted":"2019-12-05 02:09:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697286-1575511750773-ec76e134","meta_rev_id":"e39368814690d5c0d8709e67916bd753a26e89de","_number":697286,"virtual_id_number":697286,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-05 02:09:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-04 22:58:16.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-04 20:21:20.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2019-12-04 23:43:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2019-12-04 23:43:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"date":"2019-12-04 13:42:14.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-04 13:42:14.000000000","updated_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"REVIEWER"},{"updated":"2019-12-04 22:58:16.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 23:43:04.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-12-05 02:09:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"03fb5451954fc575dc497afeb9d0cda0c4cc178f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-12-04 13:42:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a941fad30080d068842f440b27a928ff5d87856e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 16:22:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d21733caeec4810a81c8b8029600269 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6f2bde90a7e8490f921b6182c9176721 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceefa976e1ae4aff8ff8e0d6246fa194 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1dc5d926cfe84ac1a5bb298405aeb235 : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59285b6b65ad4d96b0c4c5b7504afd2e : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/00ea614a385f4f44a6836913f6b07c19 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e62da53409144b1b870faf01cdfe745 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c95778714b5441e80d3704b629f7196 : SUCCESS in 7m 11s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/0537fc7bba8e4499993166ec1b963bd5 : SUCCESS in 2h 34m 03s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/4a498ca100604bf096acb7e86050e375 : SUCCESS in 52m 10s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/c6d05ed0ce7c40be9ee60f3af40fe321 : SUCCESS in 41m 11s\n- ironic-inspector-tempest-python3 https://zuul.opendev.org/t/openstack/build/25d3f8b356974e56b5812564307d184f : FAILURE in 59m 55s\n- ironic-inspector-non-standalone-tempest-python3 https://zuul.opendev.org/t/openstack/build/313b59392479454d9d2d7f5962576643 : SUCCESS in 50m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/a555ce6d023e4c3987d72f7d011d46cf : SUCCESS in 4m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8da297d9dc54f03b9d0d84763149cb6 : SUCCESS in 5m 04s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/97630c279c7443c781eddf2d973409a2 : SUCCESS in 22m 54s\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/3ce9a69b411d44949087eb18f22abf70 : SUCCESS in 4m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f07df9e53f4e6d8b00c44dc8e6582d5cc3fdb468","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-04 16:45:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/86/697286/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/046ddf7/ : FAILURE in 3h 02m 03s","accounts_in_message":[],"_revision_number":1},{"id":"6b568983bba75ebc7b448f5bf36ea3f45d9ad92d","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-12-04 20:21:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"921cdfa6d0f2d42ff2a0c9fe9310779030303661","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-04 22:58:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/86/697286/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/644b5ca/ : SUCCESS in 2h 35m 03s","accounts_in_message":[],"_revision_number":1},{"id":"346861c72de7e9610d45deba6e435b62b56ceeb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 23:04:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33a47f0d86b94fcb84f38a37bd577c3c : SUCCESS in 4m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4327c9d1eaf448e8a4009fafb44952a9 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab417ed4526f4cc9a8cedf94f06ceef9 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/74c4308a544647b592b30a2624761686 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/44d44c4c2c214c02a24eaf0519922836 : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b456d6fcd3404393aa2bc38c3269b58b : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d730920b2e04ab5946570f88da1ebb6 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5f1021f174d454997086fdc9db1c018 : SUCCESS in 5m 47s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/f7091da700cd4ee7b98cfabc1adb6c9a : SUCCESS in 2h 28m 37s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/798e872be0cf4985a77b53972aac8bb9 : SUCCESS in 54m 39s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/2010bdf49e3d4b1585b6c95cffd881f7 : SUCCESS in 37m 18s\n- ironic-inspector-tempest-python3 https://zuul.opendev.org/t/openstack/build/f5b95130f6e2458499710c605eb9eb7d : SUCCESS in 50m 27s\n- ironic-inspector-non-standalone-tempest-python3 https://zuul.opendev.org/t/openstack/build/fffdc5542d2b46c99e2f7038b1ef7012 : SUCCESS in 51m 27s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/db67fb8a91ef40cb8a0bf1e28f60ba6c : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d5e25621ae1a482585d7b2311d63afa4 : SUCCESS in 5m 14s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/01082ad279ed4f5d9b0d8bfb87cf1690 : SUCCESS in 22m 35s\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/84c4aab23f254ec2a06be46e626c3e94 : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"78ee599265125a77664760c9831586b436774671","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-04 23:43:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd46ec2bcd736109e71d5b19f8f4a49af7eeda02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 23:43:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec9e6a493198f4df8a22fb8f86a0b596297008bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 02:09:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a8dc45a0b0d4ae1af4796d1d2cf2f6d : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d71fd083d7ad4f91a238da9a7ea2a699 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/109c7d66f29241c490cb62db9d8a59f9 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec7f299f7ef14c7680e1de61d8cf3b4e : SUCCESS in 4m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1b1c7d0e92a04ed7b978e2b0bddc776f : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b330c29932b4fc0b1603c0c7008acb8 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ce93f32e9574b97b32bf1285aa2517c : SUCCESS in 4m 50s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/6dc95efafbbe40228f3cd357f8904f4f : SUCCESS in 2h 24m 46s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/dfdb2cdb42dd427e8a412f9246bba065 : SUCCESS in 53m 26s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/d6901446297d46eca5258480270cbe44 : SUCCESS in 37m 11s\n- ironic-inspector-tempest-python3 https://zuul.opendev.org/t/openstack/build/0fa481c4319847748bf3dbdeb1943fb8 : SUCCESS in 51m 27s\n- ironic-inspector-non-standalone-tempest-python3 https://zuul.opendev.org/t/openstack/build/85221a7fa17e4441a984f332ac4a83fa : SUCCESS in 53m 03s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/36211d0598d74d108a09cd9b218ee7ea : SUCCESS in 4m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a5919935f08a4b2698d7a21573391141 : SUCCESS in 4m 35s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/40bcf2673f974f55b38c7df8cf97685b : SUCCESS in 22m 15s","accounts_in_message":[],"_revision_number":1},{"id":"7e3eca2f196ffa456c599eabd437017731cc119b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 02:09:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ade8b5714b2b09c365ad7a058fc83f8a1b214a51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 02:10:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33106c1d3feb4cb0a23ab45b4e68c2d5 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd4aa4fcd80540949ec5be4c37769c63 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8521efe8e749e73dd35de7ff9109e75bcbdd829","revisions":{"f8521efe8e749e73dd35de7ff9109e75bcbdd829":{"kind":"REWORK","_number":1,"created":"2019-12-04 13:42:14.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/86/697286/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/86/697286/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/86/697286/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/86/697286/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/86/697286/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/86/697286/1"}}},"commit":{"parents":[{"commit":"3d5b57cb5de489ca0ae3912076a2eb7ea9cb4479","subject":"Stop trying to power off nodes if manage_boot is False","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/3d5b57cb5de489ca0ae3912076a2eb7ea9cb4479"}]}],"author":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2019-11-26 14:53:35.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-12-04 13:42:14.000000000","tz":0},"subject":"Get details when listing ports upon inspection of active nodes","message":"Get details when listing ports upon inspection of active nodes\n\nAdd detail\u003dTrue to make sure the inspector gets the node_uuid\nfrom the port list when inspecting active nodes,\n\nChange-Id: I1378a1a4d2a35ba8f182138877b5151f25d4c56d\nStory: #2006950\n(cherry picked from commit dc606b71320dcd96aff96cd6564387dac6d39d2c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/f8521efe8e749e73dd35de7ff9109e75bcbdd829"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/f8521efe8e749e73dd35de7ff9109e75bcbdd829"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
