)]}'
{"id":"openstack%2Fironic-inspector~697287","triplet_id":"openstack%2Fironic-inspector~stable%2Ftrain~Ifd870a8465b6643578299dc86e2eae023db6ae43","project":"openstack/ironic-inspector","branch":"stable/train","topic":"story_2006956-stable/train","hashtags":[],"change_id":"Ifd870a8465b6643578299dc86e2eae023db6ae43","subject":"Do not update/delete ports upon inspection of active nodes","status":"MERGED","created":"2019-12-04 13:42:39.000000000","updated":"2019-12-05 02:19:48.000000000","submitted":"2019-12-05 02:18:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697287-1575512296628-937fc030","meta_rev_id":"8fd580a52c94f968a8363941bb4b261f6cf0d861","_number":697287,"virtual_id_number":697287,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30107,"name":"Serena Ziviani","email":"serena.ziviani@cern.ch","username":"sziviani"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-05 02:18:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-04 23:06:52.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-04 20:21:13.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":30107,"name":"Serena Ziviani","email":"serena.ziviani@cern.ch","username":"sziviani"},{"value":2,"date":"2019-12-04 23:43:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"date":"2019-12-04 13:42:39.000000000","_account_id":30107,"name":"Serena Ziviani","email":"serena.ziviani@cern.ch","username":"sziviani"},{"value":1,"date":"2019-12-04 23:43:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30107,"name":"Serena Ziviani","email":"serena.ziviani@cern.ch","username":"sziviani"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":30107,"name":"Serena Ziviani","email":"serena.ziviani@cern.ch","username":"sziviani"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-04 13:42:39.000000000","updated_by":{"_account_id":30107,"name":"Serena Ziviani","email":"serena.ziviani@cern.ch","username":"sziviani"},"reviewer":{"_account_id":30107,"name":"Serena Ziviani","email":"serena.ziviani@cern.ch","username":"sziviani"},"state":"REVIEWER"},{"updated":"2019-12-04 23:06:52.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-12-04 23:43:19.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-12-05 02:18:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f5196398554cbcd853fcd915492b38e2e752c9bb","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-12-04 13:42:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"47fda5dc76d967d5ef1f9e32bc59b89d517bce63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 16:26:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86d62a5d308945febf6297174aba28ca : SUCCESS in 4m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20e894c6cb4f424db05dfe01b9e31e24 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d72b8fb59e6d4fdd8c4898f90b10c94e : SUCCESS in 3m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7175e87bf514682ac6ff406641da6e3 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/379639bb41f7470a9e7fa9858faee4ee : SUCCESS in 4m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/887208bd1f454e6cba94fcd1a27111f5 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d43c3c86cfd459cbf1af6d87c92d892 : SUCCESS in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4e12f7136f3457e821ea77cd3683d61 : SUCCESS in 5m 30s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/4543e724ca8a42f697026193f4979938 : SUCCESS in 2h 37m 14s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/f118efacaa0a4337898dadfad7fa88cb : SUCCESS in 53m 07s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/097991e02da94a5a812475a4a3c306b3 : SUCCESS in 47m 17s\n- ironic-inspector-tempest-python3 https://zuul.opendev.org/t/openstack/build/489e8eb1fc054b7d8e36e4bd077ba18c : SUCCESS in 58m 10s\n- ironic-inspector-non-standalone-tempest-python3 https://zuul.opendev.org/t/openstack/build/68dcce6ce8d14bafbc028ada3b412b14 : FAILURE in 57m 00s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/23757fe9e1a749f39cdb7c127889f718 : SUCCESS in 4m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0c80d64c764c42ac8b4794ac8c0f6861 : SUCCESS in 6m 04s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/778393234ae5495fa15b5c5ce973c361 : SUCCESS in 21m 29s\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/bfc6bed9e30d4abb86f4688d1d5b9b8f : SUCCESS in 5m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f02980999777440ac15f77ed447f4b9f60fe85fe","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-04 16:30:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/87/697287/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/4472383/ : SUCCESS in 2h 46m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d17aa7b046275e8936534d230dcb731db54fe0da","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-12-04 20:21:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"533ad162f2b947d100963dd5f1d62becd31dce3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 22:55:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c318785a7294d538461a135a96d36e7 : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/75f2baafdda74010bed3d56ec44f3ec5 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13e45b9efad545a6831bbc1e1d9c6e4b : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bc75c38ac9b24466bdac8442d1415bde : SUCCESS in 3m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58928890c4014f4f9023fabdf629efd2 : SUCCESS in 4m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f62ae85e6ecd4942bd54883fb496c502 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1ae9d2ad8c6459b950fe221ec792fab : SUCCESS in 10m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7133e2fbdbc74e1fb1650c6b736d4834 : SUCCESS in 5m 13s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/08d07ced6d074e3bacd17f1402524fd4 : SUCCESS in 2h 23m 08s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/33b28283b5cf47968a6696d397dc7ccc : SUCCESS in 51m 29s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/0b0957db631d42f5ab706b57ea7cdf6e : SUCCESS in 43m 42s\n- ironic-inspector-tempest-python3 https://zuul.opendev.org/t/openstack/build/dc13d4d583dc4255ae5ca9b532ee7d0f : SUCCESS in 1h 01m 54s\n- ironic-inspector-non-standalone-tempest-python3 https://zuul.opendev.org/t/openstack/build/c6534c99b7854da889957140fa78d1e5 : SUCCESS in 52m 23s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/990e669f12944488bd459cfcdfa7641e : SUCCESS in 7m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/22cc3f9f06244045b0f5329d7f0c38a0 : SUCCESS in 5m 54s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/ab337ef43d0a46d0a5fef65970f15066 : SUCCESS in 21m 25s\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/201a477ef1704e53a2de03542fc27db8 : SUCCESS in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"93ea465e8cba79011a424fded461e874fd270ce0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-12-04 23:06:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001 http://logs.rdoproject.org/87/697287/1/openstack-check/tripleo-ci-centos-7-ovb-1ctlr_1comp-featureset001/ed0370b/ : SUCCESS in 2h 43m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e3bf9aa65c15f0084790c578e2b8b5c9bf61ff03","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-12-04 23:43:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"790d50f6f6aba36e161d75cfcabc7fa32a076a68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 23:43:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffcb5b7fc3c6997f4fbfb2749ad9dde671058cd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 02:18:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/169f26e401b54a38b7233630a501723c : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b50d44a30bd43d7b7a102ec8b02031b : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6afd2cf860f44ce2a8af8aee9c707345 : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/379c64013cbd464c99fd5a4f6bf05734 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/006a560a64fb488f807c84f1e21fdc8f : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ec8c4ae1dca4a2aad5e290bf6c2723e : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c81b475811df46129612baac1b27b044 : SUCCESS in 5m 10s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/c371bb13656d448abea2676b2de96eda : SUCCESS in 2h 26m 07s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/f548ed164f0949a88e5b905939c802ad : SUCCESS in 40m 02s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/32acdfa8526e4749b94ad7e1fe5f8bd3 : SUCCESS in 35m 24s\n- ironic-inspector-tempest-python3 https://zuul.opendev.org/t/openstack/build/510586732f3047198aec94927bc7edf1 : SUCCESS in 47m 40s\n- ironic-inspector-non-standalone-tempest-python3 https://zuul.opendev.org/t/openstack/build/0bfcd390cf0148d9a74017e799a3447f : SUCCESS in 52m 09s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/9ae0dc6534954e07be9e22ec8fc98c31 : SUCCESS in 4m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/35a43b50bf744629a6bb95ee26f3f91e : SUCCESS in 4m 55s\n- bifrost-integration-tinyipa-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/c20c977a588e4e56aa14d4b492413e5c : SUCCESS in 19m 53s","accounts_in_message":[],"_revision_number":1},{"id":"bd86cdfa4612baaca24a38051ff5b47ee34c376f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 02:18:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4983214111b378ddad862e5c6a39f20b9aa97606","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 02:19:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9766d703c15745299a17e23e6d9abb14 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/850fa9b17f224f8aacd831b9268407db : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e8312c047a0b45f585ed3261457804d1f797174","revisions":{"4e8312c047a0b45f585ed3261457804d1f797174":{"kind":"REWORK","_number":1,"created":"2019-12-04 13:42:39.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/87/697287/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/87/697287/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/87/697287/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/87/697287/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/87/697287/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/87/697287/1"}}},"commit":{"parents":[{"commit":"3d5b57cb5de489ca0ae3912076a2eb7ea9cb4479","subject":"Stop trying to power off nodes if manage_boot is False","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/3d5b57cb5de489ca0ae3912076a2eb7ea9cb4479"}]}],"author":{"name":"Serena Ziviani","email":"serena.ziviani@cern.ch","date":"2019-12-02 21:59:00.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-12-04 13:42:39.000000000","tz":0},"subject":"Do not update/delete ports upon inspection of active nodes","message":"Do not update/delete ports upon inspection of active nodes\n\nShort-circuit the port update logic for active nodes to allow only\nfor the addition of new ports.\n\nStory: #2006956\nTask: #37645\n\nChange-Id: Ifd870a8465b6643578299dc86e2eae023db6ae43\n(cherry picked from commit 78dbcf603a623de16b3614187cd2c6687d2b9225)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/4e8312c047a0b45f585ed3261457804d1f797174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/4e8312c047a0b45f585ed3261457804d1f797174"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
