)]}'
{"id":"openstack%2Fironic-inspector~718613","triplet_id":"openstack%2Fironic-inspector~master~I74666983a1d4122d27a14557ba45e88418e24986","project":"openstack/ironic-inspector","branch":"master","topic":"ram","hashtags":[],"change_id":"I74666983a1d4122d27a14557ba45e88418e24986","subject":"Change default ram for tinyipa jobs","status":"MERGED","created":"2020-04-09 08:13:33.000000000","updated":"2020-04-16 15:29:18.000000000","submitted":"2020-04-16 15:22:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718613-1587050571654-2391f812","meta_rev_id":"d245e8464f8f9f1aeced2b7d654d4de3117816e4","_number":718613,"virtual_id_number":718613,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-16 15:22:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-16 13:14:28.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-04-16 09:21:55.000000000","_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2020-04-14 04:09:58.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2020-04-09 08:25:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":1,"date":"2020-04-16 07:56:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-14 04:09:58.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-04-16 07:56:31.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-04-16 13:14:28.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-16 15:22:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4d5910f0f0708c047d229fcd39cd5fd008ff2b5e","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-09 08:13:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1b5388b7bfdfe45d72bc1820341f56d7d02bbc95","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-09 08:25:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"305d1ecc9108010216147e0fed0505e59c1c7ace","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-09 10:00:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/928bb0d51a7c4316adf9186157ecd083 : FAILURE in 1h 21m 10s","accounts_in_message":[],"_revision_number":1},{"id":"d86fd0565a6427c05b631623f28db9209e883433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 11:22:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb56380a44584d94988dc81fb5b77186 : SUCCESS in 8m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/728dfcd0615147f494bcad853cd770a1 : SUCCESS in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e837bce015f4912a8c5987f6a02c1da : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db8ae0cd998a4bd8bfebd8458fe0ee9c : SUCCESS in 7m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0eb73b238051439fa5f256a47d36e9e5 : SUCCESS in 9m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8906fb9252084f578220d25da1a3a1b1 : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9317640505db415b97a3198d06e6a35d : SUCCESS in 10m 01s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/2b9ac561047f4c149c5b5b67bf84c1d9 : SUCCESS in 2h 45m 39s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/de62f0a68dde4af59f6fb9ca47a596ad : SUCCESS in 1h 09m 13s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/7f29196d8aea4c028e9cf95b5d582f8e : SUCCESS in 55m 21s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/dadd226bf3f94b4d961503a3a29d72f1 : SUCCESS in 49m 46s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/d937be30594c48138ce2ae37863d08c8 : FAILURE in 40m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6581355ed45f4767bd46deef0dfd785a : SUCCESS in 7m 58s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/362fc0c4e5b44d61bd1443f1f0344dea : SUCCESS in 20m 39s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/4ae4bdd2032646a3bd5dade24070c36a : SUCCESS in 7m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8a6d35c2c7aa5bfb2c50cdf9f5cc513e463a0c4a","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-13 21:23:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b34c0d723e9bd599d03a147c7ec6495504cd0af","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-14 00:07:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d762d479ec2d42b9b00508c4e9b505b5 : SUCCESS in 2h 41m 58s","accounts_in_message":[],"_revision_number":1},{"id":"a4d587ef8e903ec97cba7994a64286090592b318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 00:29:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db7faa34150a4169ae7e1b627925e2dd : SUCCESS in 7m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c8c660e7261410280670d95e1cc3e71 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed108f9274ee4eaead602784c4adf501 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3605421e58de443e981f88e358edef31 : SUCCESS in 7m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/116c66ece52d4e99b67b61ad2f53b837 : SUCCESS in 9m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20b3de7cf5184d89a38edab20a21545a : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f1b527406594ca89da3b2cdbea8cfab : SUCCESS in 14m 15s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/3155cb32094243beb3b84f83e43d25b1 : FAILURE in 3h 00m 12s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/1c555ce6dbc64bd3b4bb12801b79bf68 : SUCCESS in 1h 22m 05s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/77fd66e2445b4c24b2f2c1faed1db5ad : SUCCESS in 1h 05m 01s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/f587183a54a34fbaaadfbde208c5e00a : SUCCESS in 1h 13m 09s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/35d181b979ac46d29f0fecea92211184 : SUCCESS in 1h 16m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dc5b5743831d49c39d1456ac5975e30a : SUCCESS in 9m 17s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ba5425b338043298c80171b5b954c5a : SUCCESS in 26m 10s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/477449f01044464891afdefa3aba8b6f : SUCCESS in 7m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d37f434fe721ce83ffd6fe0bf4862fc6c4c6e73","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-04-14 04:09:58.000000000","message":"Patch Set 1: Code-Review+2\n\nThank you Iury","accounts_in_message":[],"_revision_number":1},{"id":"d6561ee12677db0c9463a606df1584466ca0c6b1","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-04-14 04:10:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"189e475af21a6273ada6400cedbc48aa9ed3d35d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-14 06:48:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/2a6bdfaed6db4858b802b6b140338825 : SUCCESS in 2h 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8fde3b8c8b7ef6f9c6cc79a374e15eca494f0ca4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-14 06:50:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/008efca7fd074a5aaf62623669064a47 : SUCCESS in 4m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0363884a0054e7d8b73ab2f7fb3827c : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/969d057710784151b9cb4f71239205ab : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b25881ff2664920bada31894d885ec7 : SUCCESS in 4m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7c4ea3eece1a4a59b05bb59022864bda : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ff3ac11085744a2b671d42bac267f9e : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aaf3bc9e5a7474990a1b678188099bc : SUCCESS in 7m 47s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/5795380c3a6b4d78a715981b26825716 : SUCCESS in 2h 38m 09s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/801e008396134affa2202aaa1a61f667 : FAILURE in 1h 10m 05s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/53200c076f0a49758fe2bfd0a564a721 : SUCCESS in 38m 19s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/0171b4fdbd594108b22049b793053d95 : SUCCESS in 1h 02m 14s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/41548d7d23ce4943ba0080aa58fc01bf : SUCCESS in 50m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/66f26e52f5d44db2afcc42ca8d8db601 : SUCCESS in 4m 45s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6693a0e5ca67473ba0b0a269c09a5534 : SUCCESS in 16m 42s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/30df3b338838454fa3cfe5e0cd09d04f : SUCCESS in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bd5627a23a9a796170132c53becf6b19d4954ac4","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-15 17:55:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"73924da4e7abca8b05368728e7a1db1d60f9f185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-15 20:42:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3956141ea2e74bfebf181a40a5123e2c : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f7eeaba548340e5b249a8cfa9b2dcd6 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60f3c9f8eacd4a8f86bd9eeb3390fa4b : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32a5d130b79f43159163fd759cd85dde : SUCCESS in 5m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8839327189314e988d1842039eee6021 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bce2579b637249219b790544bf325cc4 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587f106a6d1544559d5daedb74d5545c : SUCCESS in 8m 42s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a52ba96af7c94a85ab4ceb585ea1fc37 : SUCCESS in 2h 43m 30s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/a4c73eacade14bfb971c1a69f3141eaf : SUCCESS in 1h 07m 13s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/435d16d4770342bc9c59d2d54be31546 : SUCCESS in 53m 46s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/43cb1c84c62c4e69983ac495b8e51ca9 : SUCCESS in 55m 33s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/debaf285a71c44409a54c94d632274cf : SUCCESS in 56m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/76ce21fd61db46d68fbcbe3f9b462336 : SUCCESS in 7m 00s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e18e15666671453795999232540a1a4f : SUCCESS in 17m 26s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/187089ab89664f968d7d80788bd8c989 : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59f83eb424bea4b184bf5a68ac8a96a0e60f5655","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-15 20:58:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d85673eda5544195b4fcf634060fade0 : SUCCESS in 3h 00m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a6fdb0da5f750150df92bbc4f2ca442ffb61c510","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-16 07:56:31.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"138988338614e1d065ad0eed9c02a3b39db152e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 07:56:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24ee1e1c2194896c0eabd9519174762c8f9bbb0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 09:19:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16a8022cf5ca4567a07aed09e275d6bd : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9adecc8ebaad446ba76e880f34c7a60f : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1b98a68750e41d8b8af311a7e3eefb9 : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0fbef16fc90a405a800715a22bc3f878 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f85d27fde84b218c14c9ef4bd150c5 : SUCCESS in 7m 54s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/a37dade7c3664f90b936c6bf43d6a730 : FAILURE in 1h 15m 05s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/eb4562178ee74af59bbccaeca909d16d : SUCCESS in 53m 43s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/75781fab34414292907ca2e53096afa9 : SUCCESS in 1h 20m 50s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/9ddf3337ec4f4ba19c64ecdb5d08869a : SUCCESS in 1h 06m 05s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/8efcf2c180d2490c95e4c4df1024017f : SUCCESS in 52m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e7cacc6625c94eafab4da4a59e6a2153 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e13a09ba63e930b30f3d98529cfbc5e5417a7e5a","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-16 09:21:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"355db3bc87a883430bd7b0f2b77297bba4ec10c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 12:21:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6045e05366c94857a76627aef28c6343 : SUCCESS in 7m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/220f1850cc1645878c69b4d1d514bafb : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/178a02cfc74d4e38ae048aa139ce8fa1 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df5654fc8f574ceda470473b017f00a2 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53984487988848b7b57a13569997c7bb : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/407ebfce743b41e7b766f0a0588fd3c9 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95732db3ea1541c39a2d2eb6bdb4ea8d : SUCCESS in 9m 01s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/711c019fa7d246c68c4ac403292718b0 : SUCCESS in 2h 51m 56s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/eed6f8c883aa40e8afac7c2c27b4f927 : SUCCESS in 54m 07s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/90d5e12878f84902807d6f9880ae1404 : SUCCESS in 40m 33s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/d4bcab72d24c4e73952467348b53bdf6 : SUCCESS in 47m 16s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/69230b1666e64326bacf94b02d20889a : SUCCESS in 56m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87ee32f83c6c4470a8ef2c43faf6bbf4 : SUCCESS in 5m 08s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1548fb714b5d4886ab664a01859e05f1 : SUCCESS in 17m 07s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/2d217a85156249a9b23991e3cec4a584 : SUCCESS in 4m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f0315dfa4df4edf648bb6e51c6729ec8770e7cca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 12:21:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"342de0ee2483666a9620e11c21b67f31a8fdddf5","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-04-16 13:14:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/4d76690792ba4942b6cc5c1981c03704 : SUCCESS in 2h 46m 53s","accounts_in_message":[],"_revision_number":1},{"id":"a984e4bea4769c6aa439583b9971bba7d15a970f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 15:22:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f76155c8a7ab4c1abb9d4f8918492cb0 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d37e440d4743497582984ed314bb1480 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddb29014199a48a3a695d95e737ab8bb : SUCCESS in 4m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccbb4caf7c734fa2ab2a07a9c97f9f59 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c38aaf050ce349eba783c59f7d0bd004 : SUCCESS in 9m 03s\n- ironic-inspector-grenade-dsvm https://zuul.opendev.org/t/openstack/build/ea0ec21721f74614a7ddfc87f66debff : SUCCESS in 2h 52m 55s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/6924cfc73bf547b2a20ffa3b1daebdfe : SUCCESS in 56m 29s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/2d0910b72cc9443c935a2c007ee3dbe4 : SUCCESS in 39m 59s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/c8043a6daece4ff38e07f78a1d6f37d7 : SUCCESS in 49m 26s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/a542e9edd2f041eba090b29edf288e4f : SUCCESS in 53m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92fb1c709f4e4fb6a00658b21af800bb : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d2d4d3dbff6a9eea91580cd8857b717627b4416d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 15:22:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4a0cf7ed2a1825c013f08634f2c677b201e7f65b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-16 15:29:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590035736aed4080bb05eccfb57bb412 : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ebba530817356d6d4fceb6716c64eef266fe4c1","revisions":{"1ebba530817356d6d4fceb6716c64eef266fe4c1":{"kind":"REWORK","_number":1,"created":"2020-04-09 08:13:33.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/13/718613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/13/718613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/718613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/718613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/718613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/13/718613/1"}}},"commit":{"parents":[{"commit":"fed2cbfa9d01422bc71fa7877c5badf7c92ea975","subject":"Set default boot option to netboot for jobs using cirros","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/fed2cbfa9d01422bc71fa7877c5badf7c92ea975"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-09 08:12:59.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2020-04-09 08:12:59.000000000","tz":120},"subject":"Change default ram for tinyipa jobs","message":"Change default ram for tinyipa jobs\n\nChange-Id: I74666983a1d4122d27a14557ba45e88418e24986\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/1ebba530817356d6d4fceb6716c64eef266fe4c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/1ebba530817356d6d4fceb6716c64eef266fe4c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
