)]}'
{"id":"openstack%2Fironic-inspector~730516","triplet_id":"openstack%2Fironic-inspector~stable%2Fussuri~I8e46835c56995321c49af5e608b0603c25571148","project":"openstack/ironic-inspector","branch":"stable/ussuri","topic":"periodics-stable/ussuri","hashtags":[],"change_id":"I8e46835c56995321c49af5e608b0603c25571148","subject":"Stop running a periodic task for the noop PXE filter","status":"MERGED","created":"2020-05-24 16:27:18.000000000","updated":"2020-06-22 16:33:13.000000000","submitted":"2020-06-22 16:31:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730516-1592843473734-9f56d81a","meta_rev_id":"39a0ccd3975dd2062f85f28539ff316cb6f3fda1","_number":730516,"virtual_id_number":730516,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-26 04:19:11.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-22 16:31:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-22 11:13:03.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-22 12:31:30.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":2,"date":"2020-05-25 09:00:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":1,"date":"2020-05-25 09:00:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-25 09:00:13.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-05-26 04:19:11.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-06-22 11:13:03.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-06-22 16:31:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"008b38cce31883b56eccaa4774a5fe5332672faf","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-05-24 16:27:18.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e701805e910377cbb9ff835b618b6735fc5953f3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-24 16:44:36.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6e24f9eddf8246a0be9935c9b6f35841 : FAILURE in 16m 00s","accounts_in_message":[],"_revision_number":1},{"id":"600d92be95162dec13fc3b9ad1c91d885b8e5083","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 17:53:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4078902fa5b43058e3f5dc2440076a4 : SUCCESS in 6m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fb948c7ef9343abb06d01d33485e215 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d3d7d5fc02243e381131fa8b0365215 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9923a1f61c144fa392bfc89b87f91408 : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/97f43b0ab9604ff29d0811fb9e1f73ea : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a3126e34ab947aa9a2804ecfb6f9966 : SUCCESS in 8m 08s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/4e40332d252b4ad7bb897befda57673e : SUCCESS in 1h 20m 47s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/82ffb114c5834f788099980693082e14 : SUCCESS in 1h 00m 53s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/6d0424506cca4d42b203b71807527b55 : SUCCESS in 41m 31s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/44c0aee61f9f4643a21a282f7b7428ac : SUCCESS in 49m 05s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/e038f84676034406b2e3fb7524815e52 : FAILURE in 42m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c5f62c7ec2524c2bb379440be11981a9 : SUCCESS in 6m 11s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d23175268d7743ad93553fba5218aee5 : SUCCESS in 18m 27s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/0792446767bc4eccbd50ce03b4cdb379 : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a192e5bf0eb34995c6db51e9ba96411fed23ccd8","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-25 09:00:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c9348340da8573d85b4409f0322a8ebd793ca997","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-25 09:24:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ae85d82bbb9e44d09b77b8e00bcaebc8 : FAILURE in 16m 17s","accounts_in_message":[],"_revision_number":1},{"id":"5a70459e54ef23806caae5ca59eb480648506974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 10:19:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91f2339dbc6c4db49c27e8446367d83b : SUCCESS in 6m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a0818f16ce9c4c77ad7ca10bf2a86e56 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0078760574f14e219eeec72334aac126 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa5a6dff73894da7abf7b62c7186a893 : SUCCESS in 8m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/17fee11fb0404256b0f810272c42dfb6 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27694cb7802b45ac9f4ea393b366fc4a : SUCCESS in 9m 17s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/b117b4a11d9340f1b562359f8b8b784d : SUCCESS in 1h 17m 52s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/c2eb11c3a6ec49cd8d475f8cec864fdf : SUCCESS in 1h 04m 58s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/fdab88957369411b9f1fe07c240fc811 : SUCCESS in 40m 24s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/bd8cb63e07f24bf698abbac44f0aa18a : SUCCESS in 57m 07s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/9d8c51840d7b42a6bba062a726d92f4e : SUCCESS in 50m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eb0013c74d954030927aa2ad54deaf83 : SUCCESS in 6m 45s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c781e743e0c14d40b6bc4a3a026a5315 : SUCCESS in 17m 03s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/bfbf07a513aa491183f96020309115df : SUCCESS in 6m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9fc9f5d280112d7fb55d61a4ec89cfd280ee40c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 10:19:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ebc33f5d256590d82634888ce78e740b650b7acd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 13:13:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56238f7bdae8497da41981d925bbfeec : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69ea87b5383d4a3f9ccb9b9dd32610e1 : SUCCESS in 8m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b17c863ee08548409971b776ad3b7af2 : SUCCESS in 7m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e22d18932c5e46778b7e55a209830eee : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f81cd5ddf53428ab67ca4e2a8fb89a7 : SUCCESS in 9m 46s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/35d5b10ac9354375920882edaca3cae6 : SUCCESS in 1h 15m 17s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/219183ff0ec94c9a897fa4bd65dc014a : SUCCESS in 1h 08m 39s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/5ee6c5540fa24abaa3ab655dbafc02eb : SUCCESS in 40m 57s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/fa4f97f8132f41b0b4d5ceeee42d33ba : SUCCESS in 1h 03m 03s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/74f4cf548b124b4380a0bbf198142d2d : FAILURE in 1h 01m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/397460c5fc424cfba987a47097b43473 : SUCCESS in 7m 19s","accounts_in_message":[],"_revision_number":1},{"id":"09511a0dd81dc0f6f07908d65411c9d9f03f8799","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-05-25 13:26:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b9ad52d000e1aebe5c7e2e5f3eb18744c5f75f48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 14:53:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31a2228ca12d44f78be7617923d169e7 : SUCCESS in 5m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f12dc8c4e0947aba690494f455c2c56 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/021f11036602490eae40cd0229aeca11 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df152bb023064ed0a1d050bbe059ff4d : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df09613676704379b82bb9f61f98a3dc : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c02fb06f39b4209abdda53d06a2865b : SUCCESS in 10m 26s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/ce85ea96dd8c47d581472477698c2a86 : SUCCESS in 1h 21m 30s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/8ac876aabd8c4c49b4d6d6e92e46fe24 : FAILURE in 48m 03s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/514c51ba122a4d719eab84c789cb0e0a : SUCCESS in 43m 35s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/a7dd6f063f6b4fcd83ea9023a173fd11 : SUCCESS in 46m 29s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/1a7134aede0d4817bea0eb717e36f66e : SUCCESS in 54m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fef7e7171af24f4b83bb01180375a93e : SUCCESS in 6m 48s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0761f0d8ecce4ecb98e4e24dd0e649dd : SUCCESS in 16m 31s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/6964cb2ac7ce4efeab71a7362830c7a7 : SUCCESS in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5372dfb40f97193997a6b9ac1686b6a584333af2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-05-25 15:28:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7bdc01fdf0ab500200ca23a84b505b7c1d1e92b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-25 16:59:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65aff5aa04d642159fc6abafc2b6145d : SUCCESS in 5m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b394aecba084fc4a7e446d15ae70cff : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d28ca1746dbc441aa778adb5e386bc76 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98768db8954b4a3faede2697415a158a : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/64574efafcd149fb8edf84d50b9d8fb7 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/949af1c63c6a4948ac866875f437bd37 : SUCCESS in 8m 24s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/9bf96c37f93f403eb11bafb507325da5 : SUCCESS in 1h 25m 22s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/d6359afe0ac44924826d8a8c7c77b817 : FAILURE in 52m 25s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/dafcee8392364aeab203d4d92bed7b0d : SUCCESS in 38m 50s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/e1a12a5754094c18bea6468f369f5845 : SUCCESS in 55m 39s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/aa7d34d40fad4c07a330b2c781df51f6 : SUCCESS in 51m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2611f3cfdea84f07a64f14efce448a5c : SUCCESS in 5m 48s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d3aed636a084c1b950ce04255f70d98 : SUCCESS in 16m 11s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/7d11fe7e83c142d99586bdd20460c90c : SUCCESS in 5m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8bc99baa0b3a2d98221d85c23ec5126b676433fb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-25 21:38:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/ebd60db685014e8c913af64fd8843eb6 : TIMED_OUT in 4h 08m 10s","accounts_in_message":[],"_revision_number":1},{"id":"becede05e2c668049fdcfd522d50e9f015571660","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-05-26 04:19:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"92c5cbe53edeb3211e5836ff42582828720713fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 05:28:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d576c32deec4e3b80ebecee2bb1b900 : SUCCESS in 5m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a112fe32b8948309129b1f1564fc4b5 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2900d5c8116e465da83e46593096f86e : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae3bde99d8304813879b5a24fc7761e9 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5449942c09a143b6b17b154486971748 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/713219c5aa5044298c69bcbe6f83f894 : SUCCESS in 9m 03s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/69460b86da5c42eaadda5a025d48362e : SUCCESS in 1h 08m 17s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/2ddbb32cbe674d01810adf7cfeb630eb : FAILURE in 52m 31s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/88af3f15b564406e9941014bde4937e1 : SUCCESS in 37m 35s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/469da579e47b43ffb1b97850197c8168 : SUCCESS in 50m 13s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/8e0605a37d30437695ec9f76274337a9 : SUCCESS in 49m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aea5d0f8487c49ddab09204b6b2a3b87 : SUCCESS in 7m 37s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5967e8cc43904e23af7a7d160e414f4d : SUCCESS in 16m 13s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/2b937030faeb488887230525cffca75e : SUCCESS in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b281328eda68f79b0919959ef8a4e283326f9f3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-05-26 08:31:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/02d7a16cd63e4367b081b3478c360370 : TIMED_OUT in 4h 10m 36s","accounts_in_message":[],"_revision_number":1},{"id":"dc818de02b45db8db6acf98c81e49f1e1de10a60","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-06-19 13:02:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"88eed88d14db7ffb0d022cbce8247101468c4aef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 14:31:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c026fe2831ae4b62b6332becc556c6ca : SUCCESS in 7m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b51eb04282324a81a719678d38d09a0c : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21262b5d2cf442e4a36f36481346578f : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4648897e9db746c4a2814e16e6097722 : SUCCESS in 10m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a189028e230486fbd1770f4ebdeaa58 : SUCCESS in 8m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c72306dd5445cda6ed74b661e1d165 : SUCCESS in 14m 03s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/9bf822b79c5141da8b9826d89ae3d80f : SUCCESS in 1h 27m 30s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/87172911d56744a5a41ef82eb454143a : SUCCESS in 1h 05m 44s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/e8081bdb02fb49fba241d8c8a846a054 : SUCCESS in 45m 28s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/178ed6d2c6b845ffad0291510e98ebb6 : SUCCESS in 56m 09s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/c649c3572e474cd5bd86fb44ea5f13e0 : FAILURE in 51m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f14c813f6b7d450c88d53fd2f827dc67 : SUCCESS in 5m 52s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5b5418ec85941918ba3141ea670ded7 : SUCCESS in 17m 23s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/5766aa3fddb041ec812bad6e5ab7368b : SUCCESS in 4m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d70c8c3921a35a75fc5e78b1a295600785862e6","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-19 16:00:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1e25dfea73394c569264c8175f8395f5 : SUCCESS in 2h 28m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d7d52f3111ea610abfd9d4392406aca7b3098373","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-06-22 07:35:08.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2edbe996520208da888da9432183bd1c9ce1018c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 09:16:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29cebc7289424e5cae5bc69d8b17474f : SUCCESS in 6m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c503e5074d57452a803f8ee891ee24b5 : SUCCESS in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c5d1ef16cef41eba89d54e20ddac1c6 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/429720211b684ea2bf84a0555309526d : SUCCESS in 8m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f908fd40641e418aa07df243e1c8fdf6 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9029609c3ac0469b955624b45f818f26 : SUCCESS in 8m 50s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/e0e406c0df4e427cb2ca98bc15cc5c64 : SUCCESS in 1h 34m 23s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/5cc8d0ecdb3b4253a48cd424a12f7fdb : SUCCESS in 59m 59s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/ee2d946222824d8bb7a329875e824749 : SUCCESS in 44m 34s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/950ca25cf877419498ebbe3f46439a02 : SUCCESS in 50m 45s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/c73250f31dac4427a941adc8d3092de5 : SUCCESS in 46m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1101bcfbcd20425eb37a15e398d16180 : SUCCESS in 7m 17s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8b6576b85b847f39d9f98e7a501c4df : SUCCESS in 18m 19s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/16c83a631021472fa3ab29bb5d9d2f39 : SUCCESS in 6m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"28ba4b6576a4335abc566d61deaec6c1f0c4f128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 09:16:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6b5c2d85678bf95030d71ab33cbcca453a3d303","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 10:50:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55e0dca3ff9448be97c74ba78c8d47d9 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d705dad54c942e4968148d079a16223 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5947ff9d459d425f8f2c412507dff559 : SUCCESS in 6m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0622e33bcd65429dacb53a9d0eceed3a : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/561f70cb47b54eae9d40c856c441e922 : SUCCESS in 10m 07s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/93adc9a690ae4399824ee45183fdb25b : SUCCESS in 1h 27m 19s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/e82f2b91b208410ea6e96a848b2e34f3 : FAILURE in 1h 04m 07s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/791701d9c8b249aa9218a779b40d9ca1 : SUCCESS in 31m 18s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/323849f6a7be4300b622f48183a70d6f : SUCCESS in 45m 41s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/a1299de38e3f43fea54b17ab44bb4f71 : SUCCESS in 49m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1318b03a5cd340708ad3d34a8f91d02c : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":1},{"id":"5ea1f89631d663af770a703a85b011402fe36077","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-06-22 10:58:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f088184c7fecd62bc7a9421248ce1e2e7d0822bd","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-06-22 11:13:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/a1dddb0f2cf34e12bdeffcdb10fdf092 : SUCCESS in 2h 28m 21s","accounts_in_message":[],"_revision_number":1},{"id":"89f21486bf4b473b1cdb81886e225446b81cc98a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 12:23:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3174b529091f40d58de2219416986439 : SUCCESS in 5m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50cbd06634004798a0db94136b7c410c : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ea100b1ea17443f9dfd2761ccbe443e : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3651eb02599c4404aaa767a44dcb86b3 : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d6b3c284d2d45ff8273b05f7a8ec83c : SUCCESS in 8m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cdf5dcd558043319fb5438e4d7508d1 : SUCCESS in 11m 29s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/be0b1481c0bd43b09e86a574b6f17ec7 : SUCCESS in 1h 23m 30s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/fcf0e570c91d43fe80db77809be33a69 : SUCCESS in 1h 05m 15s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/4a02d76c18364cf2bc93be55ce4422d1 : SUCCESS in 40m 29s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/f3a87c3234a24b418a1c8e2e9190c373 : SUCCESS in 50m 30s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/3791dc4ff9f447c498214a705ce0b9ee : FAILURE in 44m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/53d524e1fedf4ba39daa9150e303f032 : SUCCESS in 5m 43s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c091dccf60c444d7b93bcd8dfd722c12 : SUCCESS in 17m 17s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/8cd561075dcd410b85d2402b6adf6638 : SUCCESS in 4m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"184cd613592b53744fe6acc895c4d9229e0498e9","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-06-22 12:31:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1f360a61b282d2dcf6378e1af7650c6e6b049334","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 14:01:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/060fed2595a84d778c59d238f333ef25 : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7c2b613a37f4aa5a0b1d5ee9e75d004 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1735c327a4a64a9d8d5f2971e6503061 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87ee5c20adf845f8926263c69f04cb97 : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2e3893508ba4a8884276c99d0cc5e8e : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fee972c8a6a84514877b5c42d5f9a7ce : SUCCESS in 13m 44s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/7d726388059241558f965b70c8397cb5 : SUCCESS in 1h 22m 08s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/4716329d1bd04ca0bc73d8bb385d3d16 : SUCCESS in 1h 02m 51s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/2f702e921a2547a687fe837f2553f8bb : SUCCESS in 42m 24s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/9eabae45ac8546899ba1bcaa1b14617b : SUCCESS in 52m 43s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/ac3dc6f75b6048f8a8428cd939775f81 : SUCCESS in 51m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9adb2871dc4b450fbf62136a91f9eb3b : SUCCESS in 7m 19s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5feb586f950c490d9fbb950662bda6ad : SUCCESS in 16m 38s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/ba4e31db3319496fbbe8689fcad74120 : SUCCESS in 6m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea8bba036afd611bafccb54202ec63379c585348","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 14:02:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ae7d3f3a4508340aea1d9f4aa003e32c3f57773","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:31:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"088e17698f6edf7d7ebfb8ba10208b9a45148061","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:31:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e2a2a7a2967484b92db10f35df20e40 : SUCCESS in 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e7221390e9a4b93b60a71575e51086f : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/178d6c28fa024afe97ee1662a8bc2a86 : SUCCESS in 8m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4dfa32b9b018474fb842307c534894a3 : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d11dabec4e649fba14f59106376924f : SUCCESS in 9m 30s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/7095326510f54bfeac97937b49ae19a8 : SUCCESS in 1h 29m 34s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/b08a43b57cf14a488784fb44b62788d4 : SUCCESS in 1h 02m 36s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/1c56ff4dbb9046f6a15d8a3175c3f3f9 : SUCCESS in 48m 10s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/7f57a73f020042c08fa5ccdd3a62b484 : SUCCESS in 1h 12m 01s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/67359622721c46e881d0650089ae5085 : SUCCESS in 50m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/46beb1d121164643933f585fe3753a77 : SUCCESS in 9m 39s","accounts_in_message":[],"_revision_number":1},{"id":"76f703a2e3bc1e3dd52cbf20f17e8bd518705f8b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-22 16:33:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d21bb75c8ee4951a1e68c0ca0d9c494 : SUCCESS in 1m 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"244cb9c66a2ebee19338909b7948b78c020f5f57","revisions":{"244cb9c66a2ebee19338909b7948b78c020f5f57":{"kind":"REWORK","_number":1,"created":"2020-05-24 16:27:18.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/16/730516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/16/730516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/16/730516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/16/730516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/16/730516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/16/730516/1"}}},"commit":{"parents":[{"commit":"a3e4a8bb8e1eb2c3c3255d530ca3004d25c38994","subject":"Merge \"Update local.conf example file.\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a3e4a8bb8e1eb2c3c3255d530ca3004d25c38994"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-05-06 07:27:27.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-05-24 16:27:18.000000000","tz":0},"subject":"Stop running a periodic task for the noop PXE filter","message":"Stop running a periodic task for the noop PXE filter\n\nChange-Id: I8e46835c56995321c49af5e608b0603c25571148\n(cherry picked from commit 8a02e5a538588084c40ce9651cc6d5bf330f6827)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/244cb9c66a2ebee19338909b7948b78c020f5f57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/244cb9c66a2ebee19338909b7948b78c020f5f57"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
