)]}'
{"id":"openstack%2Fironic-inspector~739813","triplet_id":"openstack%2Fironic-inspector~stable%2Fussuri~I131c8189b25ee2de66f08c742a21052ef1b266aa","project":"openstack/ironic-inspector","branch":"stable/ussuri","topic":"node-id-stable/ussuri","hashtags":[],"change_id":"I131c8189b25ee2de66f08c742a21052ef1b266aa","subject":"Use node.id instead of node.uuid in record_node","status":"MERGED","created":"2020-07-07 16:46:06.000000000","updated":"2020-07-07 22:19:38.000000000","submitted":"2020-07-07 22:17:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739813-1594160249752-4780df3a","meta_rev_id":"31bd0d8a206577cf4531c793486ec5905545cb6a","_number":739813,"virtual_id_number":739813,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-07 22:17:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-07 21:35:31.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-07 19:11:39.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2020-07-07 16:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2020-07-07 16:51:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-07 16:51:15.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-07-07 21:35:31.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-07 22:17:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f2d571d79eadb8a68aff9e319c32fe9c7dc429d9","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-07 16:46:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"777c132022eba307a52f9ac61d7c828cb5c792fe","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-07 16:51:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"246e72da94a7328506f1e93c45141e2f5e19c9ab","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-07 18:05:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/6b56516178e74bbc8ac88437d45a79c2 : FAILURE in 1h 18m 25s","accounts_in_message":[],"_revision_number":1},{"id":"e7b26d3f246a8f88b3fb5bc896e527fdb72a2289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 18:41:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb2ee3de76934891b682b47b207b414c : SUCCESS in 7m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d775c68e8dd7467594623019ffef93fb : SUCCESS in 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc7f1c97c9984d50847250da2663e061 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e884e21b1284396b056b0ec5e20be4d : SUCCESS in 10m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d319d75b1c954979a353473f34d44962 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de5e71c52c884d20bd1d98a428c8b314 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e4d15e1b97a4e62b4cebc6adc33cbbc : SUCCESS in 10m 27s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/4846d5e9b01744b4ac929fdf25cda838 : SUCCESS in 1h 06m 14s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/ea27f7cd7643454b9356324f39f3faa9 : SUCCESS in 1h 07m 21s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/8a19d7b3e34547938eda505e8b69c245 : SUCCESS in 30m 50s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/ecc2f4d3de3549dd9f32edd802ec2816 : SUCCESS in 47m 45s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/17fec67708d04b0798105810703f1a0a : FAILURE in 53m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e0192f7c05e142129056f3b2544b0d5a : SUCCESS in 7m 40s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12f8acaa345b4398bb16ce7e63345be4 : SUCCESS in 16m 29s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/fc33cb015c734aab8565d86916e05ef1 : SUCCESS in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ea450e2e273ffd81b710d515c0fe0d8a568794a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-07 19:11:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b7d0c8374d6795232455ba8daf112610de3318f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 20:44:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/274b77061711481e89f13df4052a58f3 : SUCCESS in 9m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd169d6ad2cc495280b6a96c83dcabef : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1314c1289ae249f288a78fc484c1dc6f : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59b24532235b4257b5d8858191678dd5 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cc7ba9fcbf304876abb7aa140f628962 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05bc23e3b52048789de5ce126c045ff1 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb0a24a10ac14f5a819a2752599abaec : SUCCESS in 5m 35s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/83f7ac3b988b47029e351844bc1e9915 : SUCCESS in 1h 24m 30s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/e7a5afae84384f38b3934faa94d4ec8c : SUCCESS in 57m 29s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/07b88bce0c7f4f5b99de98f965fb1390 : SUCCESS in 44m 25s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/ed9f7217871e40a78c4e53e524191e51 : SUCCESS in 54m 14s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/5245fba564bb46578174d97151168061 : SUCCESS in 54m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c97ad6f0c5e14c2d997ab334f7a4e3f7 : SUCCESS in 6m 12s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91c76df724ed41f3990924515f1e2a52 : SUCCESS in 18m 36s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/70c785bc71ae4b44b21709c9beefc654 : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb74622e811c46606fec12ba37d07055187a8dfb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 20:44:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9cead0787f075fca6707fdf008809d17fddbff9","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-07 21:35:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/fb2f29ec03074fbeb32a6704510e4e7f : SUCCESS in 2h 23m 03s","accounts_in_message":[],"_revision_number":1},{"id":"4df7864c4daf2f4c62813256ab47b3aeb818cd91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 22:17:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"18a216bd666fe680002ff9e8e2f365797f9069d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 22:17:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c2307e8524fc4921b38c0c052bbc5d75 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4f289575afd45f797062329390c5bd3 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76c7403bdd7647a9974fef468a1af275 : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/180befed622742a0ba2bf9ffdf0edf21 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b26916059beb426ead0f6f3aac1f201e : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1fb51b589104ac1a408043529d24c2d : SUCCESS in 6m 05s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/1f4c693be96b4efeb23f9ad4df1a40fd : SUCCESS in 1h 26m 30s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/6532a6297a854be98344dff81d8e8530 : SUCCESS in 57m 52s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/a67be30e873a4e36a992b2907952e194 : SUCCESS in 43m 55s\n- ironic-inspector-tempest-managed https://zuul.opendev.org/t/openstack/build/c6eea35bf266448b875273587c984a20 : SUCCESS in 54m 51s\n- ironic-inspector-non-standalone-tempest https://zuul.opendev.org/t/openstack/build/85731745e45141e0951b6629726bf30c : SUCCESS in 53m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8206c8f86a324edfa97192a8695bb7ee : SUCCESS in 7m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f072ea26270ace46140b5a8122a775476f119986","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-07 22:19:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3069bc9acbcc412283555db24776507d : SUCCESS in 1m 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26424ca396e5423e84de96596f21b3e9 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4cb904fb13ec7d1aebaefd84d64a59da9331b9c9","revisions":{"4cb904fb13ec7d1aebaefd84d64a59da9331b9c9":{"kind":"REWORK","_number":1,"created":"2020-07-07 16:46:06.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/13/739813/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/13/739813/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/739813/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/739813/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/739813/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/13/739813/1"}}},"commit":{"parents":[{"commit":"c4b29de9a0c421a58de4419cb25318545604c4fc","subject":"Merge \"Stop running a periodic task for the noop PXE filter\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/c4b29de9a0c421a58de4419cb25318545604c4fc"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-07 12:40:49.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-07-07 16:46:06.000000000","tz":0},"subject":"Use node.id instead of node.uuid in record_node","message":"Use node.id instead of node.uuid in record_node\n\nThis function does not have a compatibility shim that adds \u0027uuid\u0027.\n\nChange-Id: I131c8189b25ee2de66f08c742a21052ef1b266aa\n(cherry picked from commit 06390e372384d15448c5ed10f77d93faa494816a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/4cb904fb13ec7d1aebaefd84d64a59da9331b9c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/4cb904fb13ec7d1aebaefd84d64a59da9331b9c9"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
