)]}'
{"id":"openstack%2Fironic-inspector~740344","triplet_id":"openstack%2Fironic-inspector~master~If2e7697e3fde58ab0a4193787e29d3acdca81ebf","project":"openstack/ironic-inspector","branch":"master","topic":"740344","hashtags":[],"change_id":"If2e7697e3fde58ab0a4193787e29d3acdca81ebf","subject":"Respond so Apache HTTPd doesn\u0027t think the request failed","status":"MERGED","created":"2020-07-09 18:10:13.000000000","updated":"2020-07-24 07:32:10.000000000","submitted":"2020-07-24 07:29:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740344-1595575778790-bdb66bb5","meta_rev_id":"59f4be0d0acd218be9973c37bdc18cb982613380","_number":740344,"virtual_id_number":740344,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-07-24 05:01:48.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 07:29:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-24 07:04:33.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-07-23 08:54:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-23 22:22:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-23 22:22:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-15 23:32:06.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2020-07-21 09:58:43.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-07-23 08:54:09.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-07-23 22:22:39.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-07-24 05:01:48.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-07-24 07:04:33.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-07-24 07:29:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efe4b397c13648b0283b22973f57e242be158970","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-09 18:10:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b09f65fd278770d6183a09c8d2c9353a9a5fc559","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 20:14:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a91646fbb308480d938e66a5e7cdd851 : SUCCESS in 6m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb00e59e8d0e43e182f271d51f450d49 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99836df0941545d7af1311efb4ea89a6 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a15752bf30054cdabfbc0f45ed85ca9d : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e168ab8e34d47e084a023a47f3d2aad : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/291e7b58c5b044448b171490ec88c807 : SUCCESS in 8m 05s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/2ac79454822e40c0a0039324b3535a27 : SUCCESS in 1h 14m 35s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/0697e80dcd404d69b8356e2f52d24993 : SUCCESS in 58m 35s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/4ff94853972f447e9882cbd0d2d17496 : SUCCESS in 49m 17s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/0272392f95134603b012fd632d36fc3c : FAILURE in 1h 56m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d4f3ff61227947b68e5e59921dd848df : SUCCESS in 7m 43s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab993670a1d543a08cc6b1511f22d485 : SUCCESS in 16m 00s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/dd259729c1fc4f97b10b6146ce733d2d : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"586c96cfd00ef8e0039b0f7941bec27374460ed0","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-09 20:24:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/909acafcea704a7d853488882ec6708b : SUCCESS in 2h 12m 30s","accounts_in_message":[],"_revision_number":1},{"id":"ab6a24b357ef651e661ca8b95d3af1952be7ab0c","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-09 21:23:04.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9d10b56b8d4546adeaa5819536e484d778d5c2b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 22:46:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbd83b306b004ea88d566b9ab6e037fd : SUCCESS in 5m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65b547035b434d329013e4f23b4cff61 : FAILURE in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54453e726edc4821a137e60f35affd9d : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e6a3eec55c94681bdae702a706e8128 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/846ece0d0f734db4b767eec7d14bdbbe : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdef30dfd412492abfea0755cac9e104 : SUCCESS in 8m 33s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/7655e70273ad4c97b94c324222f8659d : SUCCESS in 1h 13m 28s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/11991f1e017d4e66b95ec72fad824797 : SUCCESS in 1h 22m 15s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/5609b09c5b8c4b36bd307c2d6dc9ce6a : SUCCESS in 48m 11s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/feadcfef1b3e4acfa1454939941d4eba : SUCCESS in 1h 22m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/59d78b187cde4ea4aa589adfc0618a95 : SUCCESS in 10m 16s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a52a7e4ee664275a6a8a0fcbdbac66c : SUCCESS in 18m 30s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/cb87bcfa9da64318bfb88d40565042c0 : SUCCESS in 8m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7f653cecd19d342485d1e6e42621c71cfeb5bdc1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-09 23:56:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/0b443c57de5e4683aa02e43ab7ffb36b : SUCCESS in 2h 32m 05s","accounts_in_message":[],"_revision_number":2},{"id":"6c40258ffc75d9646d98aa3a8821ce162bc845e8","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-10 14:34:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7b80cd0d82341892cdd091760b9fb7c49b073433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 15:43:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58c41bbaad7b47f6a3e49fc1948cea91 : SUCCESS in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37c68242e5e64069a4a68d77191a06f9 : FAILURE in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ca1d370dc474931bcd2814b9e6dfac6 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59704a8ebd3a4102ace5b8757acb1d7e : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d52159caee840808c5f9680708c4aa8 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2cb8498e30847cd8adb6e39aae13369 : SUCCESS in 10m 39s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/eda3af731b8a4b609f9ee4f9125328db : SUCCESS in 1h 04m 13s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/49cb6dae30c248c08aa9cf6a9f6b480e : SUCCESS in 1h 03m 33s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/079e20e20069475bbe3a72b310164bda : SUCCESS in 51m 13s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/e4fdfa20176e4f41897301345353675d : SUCCESS in 58m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7cf9a5d00319408d80b71839ecc0ccc0 : SUCCESS in 6m 42s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/721ff067f8444780adee39a47d74adb7 : SUCCESS in 15m 50s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/6aa8dc7b139c41baafb58d4e85a9a767 : SUCCESS in 5m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3d861d0c8a2db5e754d087e88317a79580479285","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-10 16:56:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/d1df3edfe0fd43228ab4a64718b0da2b : SUCCESS in 2h 21m 18s","accounts_in_message":[],"_revision_number":2},{"id":"5986c2a733db13d73881cfd88f41ebf38c656571","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-10 22:07:07.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"03367740a9e0d3492a00af9ae46d7ec9122993be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 23:23:08.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f91aa54cf8a4ccabb9500244756d488 : SUCCESS in 6m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c14b34aa209a41b1ae594e1cb624c7bb : FAILURE in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe13a01deb2049e0970334ace8abafaa : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4908e1467b94cd6bd6af2407b783aa6 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f2a5da12e5a847a68d6cec39a48a15b6 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd9caca348542b6a1054e1d36dd06ff : SUCCESS in 11m 50s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/ff042174a4ca4b8db8bd35e0e842c0a9 : SUCCESS in 1h 10m 16s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/7f09ecc291c546748ca79604b2d50a09 : SUCCESS in 57m 28s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/f07649731eb54ef6810510909e91a516 : SUCCESS in 45m 50s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/0a3bdb6147704ef7817d21a807319963 : FAILURE in 58m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e2222aea522f4aab892985fd54b432f7 : SUCCESS in 7m 34s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f043f02749a649719d585aa72992cc8c : SUCCESS in 15m 34s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/032ba3b651d04740b946d2bda68197db : SUCCESS in 5m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"153ce9ac3c9daf5e70393ebce5caecdc9e310c43","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-11 00:09:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/717837b4fefd44bf8bd5aec038473ff8 : SUCCESS in 2h 01m 13s","accounts_in_message":[],"_revision_number":2},{"id":"bc90735ca74c9b371479205ceae61333b09f882a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-13 18:44:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aeffb658b6c407e57e0d5c6df59eec749fa1f0a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 19:54:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e5d5db0834041dc937ced492a1f352a : SUCCESS in 10m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62d34995080041dab9050021cea9d48b : FAILURE in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7be8edd118a64989bffa1a6d708eaa6d : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9aff18a62a0a4f6e9d231a92f8f416a3 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/987c1cab9d8d4770ba261597174c0d51 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f218aaf092a049228b68293da79aa9bb : SUCCESS in 9m 06s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/a1ab398a3ca744d393028324c87b015c : SUCCESS in 1h 08m 04s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/78dc741324d04ceb95b12448b793b343 : SUCCESS in 58m 31s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/dec3d5efc4df4bb8b2d0d1118001b1b1 : SUCCESS in 49m 50s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/84676c0d805147128c207628b623d7ea : SUCCESS in 57m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1f95e70815634aa99b769b5345cec966 : FAILURE in 12m 11s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/571162607cb047d1b296f88b6ca6700e : SUCCESS in 17m 08s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/90cd412becdf4724b5c0ff87e42154b2 : SUCCESS in 6m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a21e805477a418ecf55503169a26b002ae4eb12b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-13 21:07:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/82efa6a8572e4ec09f79b1d9d0ddea53 : SUCCESS in 2h 21m 57s","accounts_in_message":[],"_revision_number":3},{"id":"ad9fe8b2a69ad3d6f83dd331fc299283d15cd455","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-14 21:51:44.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"751c67d28bf64be8c7fe2e9eed8dfacd9af8f292","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 23:17:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49ea881dae814b5b80bebf53c5763816 : SUCCESS in 6m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d64cb7c0c7d640e3a7e9c9fff84c0927 : FAILURE in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c39fd2f09c3a483e8f15f570cde3c6f1 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8af565c2becb4ddfb206f9994fc5ba57 : SUCCESS in 7m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab4252350ed64e83bf9c4cc73e753507 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65975ad269c74e1f874d605f75b96ae8 : SUCCESS in 12m 57s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/84e0a7fb79d840f28af5c6f4bf4ae11b : SUCCESS in 1h 14m 10s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/d3a100c564a34e13994dd968045f9c53 : SUCCESS in 51m 15s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/6f24a6db9b1c4da8ac1a65239d70cea5 : SUCCESS in 47m 10s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/e9886f5f3a5e4b68a8e1fe8af38fa73a : SUCCESS in 1h 24m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e49c1f9ec3a493eae2de359696af550 : FAILURE in 7m 47s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/961a3a46dc224e9683cf432e50fb6ef9 : SUCCESS in 20m 00s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/d5e02c29b757414cacbd335d3cdaea78 : SUCCESS in 6m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d2f0d2a03a5bf391e992e2b10a5416fb6c344c24","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-14 23:48:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3092f3c8774142bbadad73301cea8d1d : SUCCESS in 1h 55m 39s","accounts_in_message":[],"_revision_number":3},{"id":"4af4b9368be704851a4404171fd8e5d38ae22834","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-07-15 01:16:27.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"27f7a595e15b5f153c5468a20519d563d9aaab0f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-15 08:58:18.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4d9b169cf96f34a3a21939981551627011601f3e","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-15 14:45:47.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b59d393e8b9e8632531535e00b8adf710c60117f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-15 14:48:13.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1fb39295ca8ab9300dea8893e441b9e700ebf6ae","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-15 14:55:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e7cabb792cd6cf064213bc129fbb7c53d54988bd","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-15 14:59:52.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"01a7b2d34b39ed3d241d55b4028ef9eb84feb5b5","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-15 16:22:00.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"5b3cb20fa3371fa1054243888ce8fc18c9070af7","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-15 16:23:14.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e4c201880a85513edcd238040379469ca6c7c659","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-15 16:28:33.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"0281fd7028100b4b9adc1e9b310eef9b52c8a5c1","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-07-15 16:29:06.000000000","message":"Patch Set 6: Code-Review+2\n\nThank you Julia.","accounts_in_message":[],"_revision_number":6},{"id":"1f0b9f406db3d39a8851609e427359a39e46447b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 18:13:03.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22dca5cfd3d041bf94960876710f344b : SUCCESS in 7m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7677beba03754fd2894c1126f42d93be : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83ef4aa0ead8405caad8e31a1659cca1 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54dc0932a0f84dbc8160488bc58186ad : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b24b063d8c60423a8cac23fe5e13c2ab : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/914cd8562ee049b48d10710b673509c5 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2d5139daec345b79a797517ae1dc0a4 : SUCCESS in 6m 00s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/653fd1300f574d888862c823b3e7d2a7 : SUCCESS in 1h 22m 04s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/ccf9fa80d9044b4a8832f2e4cf88c382 : SUCCESS in 1h 01m 14s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/7d66df806fd8480981c633f578f3fe38 : SUCCESS in 50m 13s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/326c517d3c7a42868a78e64123247d5d : FAILURE in 58m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d1d884ed5ca24b6faf9dfa316389caae : SUCCESS in 6m 15s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fc29db77145465a8520b79e8c43e83a : SUCCESS in 17m 48s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/cf60c1a9192c49bea19834d86a0eda3b : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5cdb1c0d02cdba42e076269cba3720361acdbf03","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-15 18:26:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/effb422878dd4f4fbf5fb90a914c04df : SUCCESS in 2h 02m 26s","accounts_in_message":[],"_revision_number":6},{"id":"750981db76e9fb162ba002accbc43303c55368b2","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-15 18:41:06.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"14a2f977d61048af416392e8da763370d15040b4","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-15 20:39:14.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/dd279cda9f7b48888e950f76e51155ac : SUCCESS in 1h 57m 13s","accounts_in_message":[],"_revision_number":6},{"id":"5ef89f288ff48300c35051f7985dafc51d0bd421","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-15 20:39:40.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbad2d497046407498ca60fbfda91599 : SUCCESS in 6m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f55e3eaf852a411083504da39ed7ae2e : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a56b09c97cc4117a5088b380f9786d7 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd28f3dc4dee4b3ab08d724fc17dd02e : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5feb1ca4104e4c6e8a3744be705caabc : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10bc33e39ac84a618650a372231df317 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbf97a73e29744a6a3e92fab83180d3a : SUCCESS in 7m 58s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/ca5ff65edc064d0e8095dc32776e24ec : SUCCESS in 1h 21m 30s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/27a85da9e2bb474ca9ed4529fc4ac3fa : SUCCESS in 1h 02m 18s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/bbc7128eda834d51b0f142e1b96db7a3 : SUCCESS in 47m 22s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/999d9074974d41e1898ddf0ef59f5b26 : FAILURE in 1h 54m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/072ecad8e4784cbea2a80def3bc6c84a : SUCCESS in 7m 07s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/263cd7bbdd4542b1b0a00974999d2f9f : SUCCESS in 17m 51s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/7915c822c8f4496ebd0c817d5a845f52 : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3d3edadf505df0f76f3045d6c8effd129ca7adc0","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-07-15 23:32:06.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c5b7b37bbd3b3ff9b79b255276979ce0d7d24954","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-07-16 07:00:56.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"1965d755fbe11ac5fb243060a8881a263899f141","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 08:18:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52563ada348b4f2a9abf1abb94299d9e : SUCCESS in 7m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/383d1febab7341ca9e54574adff97324 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7439bc2d49f14f2f81641084a40a6dde : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11cb9c3d90f84ce192feb745255ddeab : SUCCESS in 8m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c58e8dc22d747a6ae12b86929903ee6 : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b85cc98e5bf47938b3a7bedce37ff1e : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4e970aa50224c0b8e30ca23aea60ab3 : SUCCESS in 6m 41s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/4decf452aa9445be9a95c05e750e0002 : SUCCESS in 1h 11m 08s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/48e39c4585a143cbbdfd23dff1205fd6 : SUCCESS in 1h 04m 25s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/26cd45abf12d41a694dd12c9559244d2 : SUCCESS in 35m 01s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/5b5cc1dc170e4335bda03858ae21c7f5 : SUCCESS in 45m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f61a9675f69140ab920d2e5f5b28743d : SUCCESS in 7m 12s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9c04d586b4b46d3992e1ad9a8a779d8 : SUCCESS in 17m 59s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/1c19bfbf8344437f9a0e3638923f05a1 : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f7669813169bd9af75f581366aa2bf1812ea37b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 08:19:08.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"4cdf3164975da03533ade76e719bb30443e84fad","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-16 09:01:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/3e977d3f934b4c6eb14319aa608bc5af : SUCCESS in 1h 59m 40s","accounts_in_message":[],"_revision_number":6},{"id":"1cbad687362051c1dada7bc84b01cbde228e1159","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 10:34:28.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/464094b0244c4e158667c229c5a2ee76 : SUCCESS in 8m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd9640adb22476ba0a0bb165a96809c : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da7e8b10115140d18bce574796d3d7ee : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/105b6110a70446be87b1f6974f47b6d3 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9740515b4ba24fffa53a9937282ce6d4 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e03ced2388604a62acd3358f3f74160e : SUCCESS in 5m 26s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/e0335b561f454586a983178e10f324fa : SUCCESS in 1h 20m 43s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/e23ebb450f074b5e9cbd57c023ef1ed3 : SUCCESS in 1h 06m 20s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/ba2fe0f759714043a2cc6692936d5048 : SUCCESS in 1h 00m 13s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/29531ff158ec4bbba2298a5d5c43a019 : FAILURE in 1h 56m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f66aea5220d54325880f0136328ec1c7 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":6},{"id":"23594865072c4346e2aae01b6b8483c5090c1c89","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-16 10:36:57.000000000","message":"Patch Set 6: -Code-Review\n\nThis change failed because of the problem it was supposed to fix :( https://zuul.opendev.org/t/openstack/build/29531ff158ec4bbba2298a5d5c43a019/log/controller/logs/screen-ir-cond.txt?severity\u003d2#1019","accounts_in_message":[],"_revision_number":6},{"id":"d0e6b7522b577a21c8f3f22a18de1c6aecddf56a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-16 13:03:42.000000000","message":"Patch Set 6:\n\n\u003e This change failed because of the problem it was supposed to fix :(\n \u003e https://zuul.opendev.org/t/openstack/build/29531ff158ec4bbba2298a5d5c43a019/log/controller/logs/screen-ir-cond.txt?severity\u003d2#1019\n\nMaybe we do need to send content? 202 doesn\u0027t forbid sending a body payload...","accounts_in_message":[],"_revision_number":6},{"id":"0fe7b03af84ef34c0641a5ba2250937779ba2c44","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-16 16:46:15.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5d60df8d8ca160bf1124cd23483aeb9d410f4a94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 18:04:20.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74cfb4935c424ec6bcb3a8a087e666c3 : SUCCESS in 5m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6041795e7684448a931b1dc00ef08f6e : FAILURE in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3db083ddc5049aebf61d5cf51863329 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddd710889f0d482d8b2b5f23fa7002ae : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b85be0a68f25486e98898e5a3623df38 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033a36e843e4460694e87a269a8c3743 : SUCCESS in 9m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d7898719b6349a1a88002d1a914772e : SUCCESS in 5m 46s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/5dbe3dc8fa7841e4b6f87d24042c1f78 : SUCCESS in 1h 08m 27s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/015b4069fc884e9697b09385b74b176a : SUCCESS in 1h 05m 15s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/d881ddc7331e439ab555102db54c16fa : SUCCESS in 53m 59s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/24954c1b73da402988a567ab928d30ca : SUCCESS in 1h 04m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/74d2142278444a9780651383b2ef303a : FAILURE in 7m 21s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efebce9ceeb04bb4916efa872c157679 : SUCCESS in 16m 50s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/1517e82209b841c58c571789ba50ec20 : SUCCESS in 5m 06s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6a5fca75566806afa4d2ed8dbf619c4a7048ae8b","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-16 18:50:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e9b6db5fb9e24b909ab6df97a3206083 : SUCCESS in 2h 02m 44s","accounts_in_message":[],"_revision_number":7},{"id":"60c4ccc37277b946b8d77f9c1df5d2433d9193e8","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-16 19:38:22.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8cddf8bd7ec9d95b5530c8b85e65ffaa58bd83c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 20:56:05.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/47b36f7934af429fb6b97cbb706b0f51 : FAILURE in 4m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de4a08a0f32f480c9d00a2487f2a260b : SUCCESS in 7m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fba4c35082449c1831a5cf00838c05b : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b746b14cd8a47e88e4aa9558092cf11 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c033125bd5ea4359a16a118f4d57cb4a : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4124fb1d5c34574a60093875bdc59d3 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade562195d40414cbe703015763c776a : SUCCESS in 10m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f1485d36cfa447af891337bcfb3aafee : SUCCESS in 5m 23s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/edf2a921d5834669ae62e70becdb6783 : SUCCESS in 1h 11m 49s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/b492016f78d244abbd2591832fd6347a : SUCCESS in 1h 00m 06s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/7f89f0a51ced42d6a5ad2a8f5003d1ef : SUCCESS in 45m 57s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/6a5feec6933a4adeade58f089325c2c9 : SUCCESS in 58m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b02f8b934bf44c29b0b358b16d9671ac : FAILURE in 6m 11s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94e039b91cb34878a92f5e23c0f25140 : SUCCESS in 15m 55s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/2cb703fae1ae4bff9c27df614f83351f : SUCCESS in 5m 31s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"652b8e82b68e065386a0f1c3c807f8e07b89ba7d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-16 21:39:04.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/41e20ed023b44f029b7c3023f937e174 : SUCCESS in 1h 59m 46s","accounts_in_message":[],"_revision_number":8},{"id":"156da8325f0e1d25647709064a129ebfe064b9a3","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-17 04:22:16.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"f2bc1b059e1030ff0ca14d56a03fb5ab829945ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 05:37:33.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/08d6b900ab50495ab7fbf428367d0ce0 : FAILURE in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57a93598bcc54a999ed240dabbd5f8e6 : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0721a866848340de9634a054a1bc5a82 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d959e4f0c52f449b9eacfddded2a900e : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f69c1fddeee46f7bf6f2a444c34085d : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c247a4a5b8304d5581fb17f3f8f37000 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d610277c1a854a14a2c2f852c4cd79f0 : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a0f73c721784ad19040c5f5f73c29e4 : SUCCESS in 5m 41s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/0a62652a26cb4ef290b2bb7c3d33f659 : SUCCESS in 1h 13m 33s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/723fe225f59841a48fe938c32d4cb9ea : SUCCESS in 52m 45s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/15083ea7711a4eb5a88b39ab81f8817e : SUCCESS in 50m 08s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/1a5f8cfdbd47423086eb0bfd89024e43 : SUCCESS in 1h 07m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4eaded932d2240c5a8c88fc3ed99fbe7 : FAILURE in 11m 38s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20315ee2938742e396c748d7be97c0e3 : SUCCESS in 14m 47s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/9e129ae6ceea40faba4b295beeaae337 : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"446b38e898e2d62d5d3f675798f05b10dcf47ba6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-17 06:20:42.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"7174252859fcd1a7057d1c25d956d5c5955eeebb","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-17 06:25:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/084451f5a5c44e0387afba3f07d8a55b : SUCCESS in 2h 02m 19s","accounts_in_message":[],"_revision_number":8},{"id":"5eddc00df7556ff04bfa187981e5751822aad40b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 08:31:56.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/9f73a68fecec4118a40c96234912e1f0 : FAILURE in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0951998b504470aa0e10dcd910850bb : SUCCESS in 8m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab3c97e2803e4f13bddfac67142cc378 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bdbfd105e114649ae7a8d9cc47a93d7 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c5140e763d14b2f9d0e56a86c9de705 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23a1b076c10041babfa7a3403df062c3 : SUCCESS in 8m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b9e7991a6aa41ac924e298f094faa1e : SUCCESS in 9m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da309c1304514c5a936ca497cb217f2d : SUCCESS in 8m 36s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/dd79756ef4504d7394bcf06d22ae295a : SUCCESS in 1h 16m 48s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/74c8b79cb1d041c1ba0ac9a2f4e4c9cf : SUCCESS in 1h 10m 50s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/0f2a5dde87ea4725aa65234c89329833 : SUCCESS in 52m 19s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/cf9e627241064ca7adbd6cb78ba5d9b8 : SUCCESS in 1h 11m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db670661946e40c1a194a51c5e83b40a : FAILURE in 9m 07s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c0963bcf4d7459a9c5ad6f1c6aa3fe7 : SUCCESS in 16m 23s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/a391cb0585544e02a520d48c856fef88 : SUCCESS in 6m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"96d8163e5da6fafc5d36304a39b5ec968df9a05d","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-17 17:59:53.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"8b403105fef3ec63f6a8efd48fd3dd98ef0f5e22","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-17 18:00:57.000000000","message":"Patch Set 8:\n\nFour successful runs... so it is a start.","accounts_in_message":[],"_revision_number":8},{"id":"acb4f9070e77f073652570d5ba22db980315157e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 19:24:54.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5f1a15c2a1b64fe4b55b1c307b5ce59a : FAILURE in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/028c0e42671047e5a893c01da7ea2570 : SUCCESS in 6m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c6e3236078ed440a8a7374267aa9abc4 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05a57f30acc242e58a44ebbf5d7cc374 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4956cbbbacc485baa915a57271dc4a1 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1801d298ff34936846462c654082f42 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b749d8a2dad42f09d9a124f7a2d2200 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92826d644acb4f6fa5962b734c1c7a76 : SUCCESS in 7m 27s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/ff0f4448026348c3a7fdcdf6808f3e13 : SUCCESS in 1h 19m 57s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/39e3b3f9c0fc4eb8b5d5ea5c475d4755 : SUCCESS in 1h 03m 35s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/880cf61d8b9f498e8128f1e7376a6e7c : SUCCESS in 51m 20s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/85c7d5d4d56641a4a60d89d6964060d3 : SUCCESS in 1h 04m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db4ce2f2a5c24503848b6be980466d71 : FAILURE in 7m 09s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91558d8b9b8b471985cbd88d6e966819 : SUCCESS in 14m 54s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/286be9deda2642059c92a9a5683a5b87 : SUCCESS in 7m 50s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"75cd0d2bec9e7bf5973276097aa0d81c64bcc4ce","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-17 20:01:53.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/244371a59fca4d3fa94d38ddcaf2e835 : SUCCESS in 2h 01m 14s","accounts_in_message":[],"_revision_number":8},{"id":"c2ea428a2cab9649448d8358982b4386a59e62e8","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-17 23:19:59.000000000","message":"Patch Set 8:\n\nrecheck five...","accounts_in_message":[],"_revision_number":8},{"id":"b8f3f534241cc733231fb1f4aea0961aefa2d45c","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-18 01:29:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/fdb823a05ec34cdf888e2835bbe85c21 : SUCCESS in 2h 08m 46s","accounts_in_message":[],"_revision_number":8},{"id":"9ecd25a2052bb216b35223f2a81a2c325e94377a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 01:35:29.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d36692e0c5534d8394f590d83386c052 : FAILURE in 3m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ccbfef19f3e4056ac11513aaf93ef14 : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/939e65e6d37647ab83b7144b2486da99 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca91c17cedc84a3d8df6fb28b2715a7b : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73fe92d25b3e4616bf05b7efef34cff8 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d3daf1d4196484a9ee036c92a4297d1 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11959bdbda4f4a64b22390d2a073b3c1 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3310c0e56ce64a70a0984dec3cd702d2 : SUCCESS in 5m 36s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/705a58b4a0274ef19ad7be79c0799eae : SUCCESS in 1h 09m 41s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/9eff716007374e108d4c18fcee748db7 : SUCCESS in 1h 00m 24s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/f91e735238d84674b3b92060beb6759c : SUCCESS in 46m 23s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/490754c5b474425d9b20764c45d70893 : SUCCESS in 1h 03m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7295b3915dd14194bb96e9d34d33cdbb : FAILURE in 5m 31s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf5144e08a0e4470b7800e1105fd3d71 : SUCCESS in 23m 35s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/eb6b63160f224899ae6f6dd9cc7d855e : SUCCESS in 5m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"403b2e9986ee32fbac3ee3c66f67e5d040181aea","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-18 19:37:12.000000000","message":"Patch Set 8:\n\nrecheck again!","accounts_in_message":[],"_revision_number":8},{"id":"52f092a64aeb745833b906f6e0264234664f9be6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 20:46:00.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/53c777f2bb0e40b1b1a7578c77ddf897 : FAILURE in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e32492a337842509bf4c2d2b251f88e : SUCCESS in 5m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9d5b7fd346bd4788a8a88d8436ac347a : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16bf3754c305405587bf90d4a5788525 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cbed4eef8a34444a0840bbef4166e4d : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d45ad5519ed24d07ac60b7091c2a37f7 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0701702ec234b09b1e42cae65db74af : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0371c4c4f914d669482b90fff7c48a6 : SUCCESS in 6m 29s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/00ddc2a4da014d33b5e0b590dbfbe60e : SUCCESS in 1h 06m 45s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/765d63d735a340fbac76dd85f8f10a20 : SUCCESS in 39m 22s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/4f02b111399d405aaa70152c5c0596e7 : SUCCESS in 43m 06s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/4cdec35bbe904be2ac2d05a14481e379 : SUCCESS in 55m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/454dadac7cc547a38995ab94f54aa688 : FAILURE in 5m 14s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21d16c31750246b6bccc739bd9731a77 : SUCCESS in 15m 03s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/74c4a501203544f49193fbc4f27535ef : SUCCESS in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b32e239b8b78e6c9ff26c81a984decf0c383c3f1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-18 21:56:33.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/03ad0392ee7e487f881b06c604befbe5 : SUCCESS in 2h 18m 31s","accounts_in_message":[],"_revision_number":8},{"id":"c30cdc507e3aea99ba7541581fd2ae15a86be8da","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-20 22:46:53.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"7ceca84fd998aa467a4ac6edc58f0a7bc5943991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 00:11:48.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b9316d26dcdf4d9cbcbd023cd4e13ca0 : FAILURE in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6a525ba5ead4877a7da67844ea93fdf : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fda34fb26bae4c6b9536efedc1fdd3b3 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f32086665944648ab093e09b946120b : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fdc05999efc49e9a1f6a5f8706daf50 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9666d38ac85d4fe6b3f5f7987cbbc9de : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beb426f9afc541a3af43097b2b5572d0 : SUCCESS in 14m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb305b6ff4dd49eca308afd86cd135d7 : SUCCESS in 5m 02s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/52fa0906b44d4d0a9107d00bb7e77dd0 : SUCCESS in 1h 19m 25s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/d1890fd999b34b2cba6ee55981211094 : SUCCESS in 1h 03m 09s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/93df1e1b213347be895eb2cd9300ab4a : SUCCESS in 37m 50s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/0c3a16c14cdf4829bd066b2808f59d29 : SUCCESS in 40m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/59ce666909a34b3aa7c6b7cc7a1b52ab : FAILURE in 5m 20s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f1185939e10408783579b8cead0c7a5 : SUCCESS in 16m 18s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/c9e4eaf6fd0f430c9274a8307b3691b6 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0ec09411e4bd507a3eb99a999d8dbd0e7fed5b10","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-21 00:49:00.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/5356a0502f044d0aa8a530fc12e0433d : SUCCESS in 2h 01m 19s","accounts_in_message":[],"_revision_number":8},{"id":"70d7028df13fe747ff4ce0b98ea0a8c16068f47e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-21 09:58:43.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)\n\nNice! Just two comments inline.","accounts_in_message":[],"_revision_number":8},{"id":"07c283e1d073cc9ac50567102335417458fb4656","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-07-21 09:59:16.000000000","message":"Patch Set 8:\n\nOh, and the functional tests need updating: https://zuul.opendev.org/t/openstack/build/59ce666909a34b3aa7c6b7cc7a1b52ab/log/job-output.txt#1529","accounts_in_message":[],"_revision_number":8},{"id":"c8b8034903ba8c3ad88074a18d6f1aad4d477081","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-21 19:59:48.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bb3fa56e03d81ba7ceddd7300283281e9e76b6e3","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-21 20:20:56.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"3b3a917f6ea67b9252135e738be75943fe0fd403","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-21 22:23:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/e6041901470343a9bf53784e7b5b2935 : SUCCESS in 2h 01m 59s","accounts_in_message":[],"_revision_number":9},{"id":"7cbcc814d2c85ee0c3746f3cb6b0c460bfc61981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 23:19:33.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/63d08dc6a21f4db8aabafe51fbda01bd : SUCCESS in 3m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac5536389dc2460ebefb7818d4251ee5 : SUCCESS in 5m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7cf313467bba4b4e914efe69de5cdb35 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e5a43c8a8c4347994ab2dcbb9a6217 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2930e48bff245da81965aa6cc30f485 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca5ba919c820422c9d27273d9ff4abdb : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5ed83bf9204c619fb205c4b75ddf47 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e93a0efa83ac4b34bd5ad074bff5e558 : SUCCESS in 6m 18s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/8ad9d68085414e419be6934d301c5b3a : SUCCESS in 1h 15m 10s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/9d17bb7eb8e344048a26799ecec4f33f : SUCCESS in 1h 05m 17s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/2e9e49188f7e448dabde763caa559128 : SUCCESS in 49m 28s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/0ea7412a8e5e42869ac7b36ecfdb7880 : FAILURE in 1h 44m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d645367e39ca4633b13b6022788cc9d3 : FAILURE in 6m 48s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c5729042c474ade878a846e32b2cf6e : SUCCESS in 18m 14s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/d46e60d8bdde4636855ffd1b52eb0bba : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9f3a9640a9fb8fd6dd1dd896cd20ffbd6df44032","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-07-22 17:18:09.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b94966d5af12d9d439b614fac91738d08e367474","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-22 19:23:08.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/98cb0a7ea2a64a0abbf13b55c18a0a41 : SUCCESS in 2h 03m 57s","accounts_in_message":[],"_revision_number":10},{"id":"0e1cfdadbb9a763022c15aaaf06c8e7709196645","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-22 19:56:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3ed1b730d132454d9942edd2746b8270 : SUCCESS in 2m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/785ecba8c864461a8c7bdd4b80eae831 : SUCCESS in 5m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3dd92c2204a4cc8b256dbc47a7cbf27 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f1877c5193043628fa7b8568b239713 : SUCCESS in 8m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2d141a97e0f41d88776d9648f0962ac : SUCCESS in 8m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1f886730bed4390aece7a092651838f : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97cdb4f5fb384e218794413f4c84a650 : SUCCESS in 7m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0e94c7fb5c74d269c78c275da4e7f33 : SUCCESS in 6m 37s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/c4e1680b1700448180c05c6a22050b1e : SUCCESS in 1h 07m 19s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/3aee6f2df05c41f3be1406174b19a6dd : SUCCESS in 1h 05m 58s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/a90a47fe64724829aae584c0cbe6f1e3 : SUCCESS in 55m 05s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/27ab1287709c4d65b6e5481d041cf96b : SUCCESS in 1h 09m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4f71a21836d8427084c1c1b620e001b5 : SUCCESS in 7m 24s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32353ec49239416bbe1f8f774566a75c : SUCCESS in 18m 53s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/23c1b524da584a5284d26f3477dd4799 : SUCCESS in 4m 46s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"fc565873a13969542c6bd60f3c563cb20baa9442","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-07-23 08:54:09.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"2e21f6a5eb8804a95c6e0d128cbbb9a59d0caddf","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-07-23 22:22:39.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"820e159388cc9674144085aadd0fdc924d08f16e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 22:22:50.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"91846f20974f4f98552a40bdd7a16f78dffd55d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 23:43:07.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/c95750e42aec42ac8cd3cab8d3b30fc6 : SUCCESS in 3m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed240a622dd148b48c23c063c5d15222 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2b37d1efe8c4fee964e128d323554f4 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a3ad4c485b242738a26d8c1b52cca60 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc9bb58278f746d79a029eaf90f23cbf : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aaf15af44b3487a97d736fd016c4d80 : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/245cc8217e434a069b028198541e4dbb : SUCCESS in 5m 21s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/b9591026d8ae40a8a7e503c49026797f : SUCCESS in 1h 14m 31s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/ca868e7e4e09432eb19138bd3cb0c0de : SUCCESS in 1h 09m 03s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/899b4c2bc1d940d4b5b07687b8a28d83 : SUCCESS in 43m 58s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/f47cf7ab417543c58c3d3659e81647b7 : FAILURE in 58m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e471aecca613454590c09c0eab0f4d6a : SUCCESS in 6m 09s","accounts_in_message":[],"_revision_number":10},{"id":"8c9bca2b2dfbe00539abbe502f275081c3201ad0","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-07-24 05:01:48.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"aa96ce174ecf2af382e5ee0c5b52f7b31e80cec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 06:12:08.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ffdf8efa2f374257ab50f94d2fc5a682 : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a57cd4c0b0eb4731b3f0d7f8ce2e9a54 : SUCCESS in 4m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d16de6a025de4ad79f6dc622a3ec97c6 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7dabeeed99ef4d3891ab3a5cfdf8ce49 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aba6039bd15945e9b5ffd14f8281ab41 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/859319ad277747818b4d1454d5e5fd87 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016e7690f32f45ad8865cefff5669679 : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76751c3a187b45c1ba4d725cec25c025 : SUCCESS in 5m 39s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/9e7d79199eda430caa86e5df3e09ce1a : SUCCESS in 1h 05m 07s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/45d20eaaba81452cb6a77a3e05c554b1 : SUCCESS in 44m 02s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/749102c52d684458a9e637857ce45ef8 : SUCCESS in 44m 57s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/7f33ed931a644ab5a1d883d89fb6ca5f : SUCCESS in 1h 04m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97af05851aca4015a7f85a92ea751194 : SUCCESS in 6m 19s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44c70569b8ee4d1bb6bd9f31e5c1435a : SUCCESS in 19m 02s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/33200c21388e4d9daad72787118dff6b : SUCCESS in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b6f43d4010a72244e9889b28df317022f836ca5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 06:12:21.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"dafcc796599f3760672081964ce060ec4a0191bc","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-07-24 07:04:33.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/8af6e69b412e45428f5ff14b8a11d3be : SUCCESS in 2h 01m 56s","accounts_in_message":[],"_revision_number":10},{"id":"c5436a210757c29a7b2fde5d927aaf2db2891d22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 07:29:38.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/381e6fb2f5814ccfbf8ad2db84097cac : SUCCESS in 6m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a24b134dc4c142069c259ac7da071a43 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff52e0f930f4b2cbf44350feda43f5a : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e400fe282707406790c4d24e7cf32a3d : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17513e5443204175bb549b17fa881a4d : SUCCESS in 8m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b95cc770903b486a90e6cb9e2d5b2862 : SUCCESS in 9m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c48237d1f13433aa53a1f801db8dcdb : SUCCESS in 7m 51s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/2e49bdf68a4f42699ed94c67c527a4ad : SUCCESS in 1h 11m 46s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/0f6d0a1aa43c457f887dd5c8f1cd1d81 : SUCCESS in 1h 01m 01s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/12888c8781ea4b3194f50cc9b88c96e3 : SUCCESS in 54m 26s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/770730f4fe2a42139c8c030c2aab82ff : SUCCESS in 1h 09m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8e3755c6c3d648c8be672dce37bdc0bb : SUCCESS in 8m 25s","accounts_in_message":[],"_revision_number":10},{"id":"9451fec578d425cf531bf5c0fb81b7f2b2508767","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 07:29:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"e6f5577c29aa0274138c8fe2aca4a228f90ee06b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 07:32:10.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cf8ae99d69b4751a8c0506bbecea7ed : SUCCESS in 1m 36s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b80d3ec353bf4b68a58ef8ac71ea868f : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"6e2cb60e778180a40418dbc2011fa25763c32363","revisions":{"4f050546f5a10f706ff2b2c562c452f32c424e92":{"kind":"REWORK","_number":1,"created":"2020-07-09 18:10:13.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/44/740344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/44/740344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/1"}}},"commit":{"parents":[{"commit":"a1f8926cd161b770bbd189d86dd915b3aec1b274","subject":"Remove non-inclusive language in pxe filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a1f8926cd161b770bbd189d86dd915b3aec1b274"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-09 18:07:18.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-09 18:07:18.000000000","tz":-420},"subject":"Additional debugging around inspection start","message":"Additional debugging around inspection start\n\nWe seem to be having intermittent inspection errors with a Bad Gateway\nerror message, but digging deeper we seem to also be getting at least\nthe initial cache population step into Inspector before all trace ends,\nwhich suggests something errored and raised an unexpected exception.\n\nIssue is, we can\u0027t tell where this is occuring, so we need a little\nmore in the way of debugging and exception catching.\n\nChange-Id: If2e7697e3fde58ab0a4193787e29d3acdca81ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/4f050546f5a10f706ff2b2c562c452f32c424e92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/4f050546f5a10f706ff2b2c562c452f32c424e92"}]},"branch":"refs/heads/master"},"1c4e42e5505ef2c6e020baa7aa0e20dc44dbd39a":{"kind":"REWORK","_number":2,"created":"2020-07-09 21:23:04.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/44/740344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/44/740344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/2"}}},"commit":{"parents":[{"commit":"a1f8926cd161b770bbd189d86dd915b3aec1b274","subject":"Remove non-inclusive language in pxe filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a1f8926cd161b770bbd189d86dd915b3aec1b274"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-09 18:07:18.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-09 21:22:54.000000000","tz":-420},"subject":"WIP: Additional debugging around inspection start","message":"WIP: Additional debugging around inspection start\n\nWe seem to be having intermittent inspection errors with a Bad Gateway\nerror message, but digging deeper we seem to also be getting at least\nthe initial cache population step into Inspector before all trace ends,\nwhich suggests something errored and raised an unexpected exception.\n\nIssue is, we can\u0027t tell where this is occuring, so we need a little\nmore in the way of debugging and exception catching.\n\nChange-Id: If2e7697e3fde58ab0a4193787e29d3acdca81ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/1c4e42e5505ef2c6e020baa7aa0e20dc44dbd39a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/1c4e42e5505ef2c6e020baa7aa0e20dc44dbd39a"}]},"branch":"refs/heads/master"},"a130bc4365bf7e8867e382f80764f8381d37356e":{"kind":"REWORK","_number":3,"created":"2020-07-13 18:44:53.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/44/740344/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/44/740344/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/3"}}},"commit":{"parents":[{"commit":"a1f8926cd161b770bbd189d86dd915b3aec1b274","subject":"Remove non-inclusive language in pxe filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a1f8926cd161b770bbd189d86dd915b3aec1b274"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-09 18:07:18.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-13 18:44:48.000000000","tz":-420},"subject":"Respond so a ContentType is included","message":"Respond so a ContentType is included\n\nWhen sending a literal empty response, Flask does not include a\nContentType in the response. While in many cases, we don\u0027t need\nneed a ContentType nor expect one on the API client, Apache\nwebserver can treat this as an error and generate an Error\nindicating a Bad Gateway.\n\nHowever, when this Bad Gateway error occurs, the API/Conductor\nwere proceeding like there was no issue. The API client on the\nother hand thinks that a hard failure has occured.\n\nAlso adds some additional catches to provide additional logging\nwhich turned out not to be needed in this case, but it would be\nuseful for others.\n\nChange-Id: If2e7697e3fde58ab0a4193787e29d3acdca81ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a130bc4365bf7e8867e382f80764f8381d37356e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a130bc4365bf7e8867e382f80764f8381d37356e"}]},"branch":"refs/heads/master"},"06457f154552a74ab2e373fa35be210b05803364":{"kind":"REWORK","_number":4,"created":"2020-07-15 14:45:47.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/44/740344/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/44/740344/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/4"}}},"commit":{"parents":[{"commit":"a1f8926cd161b770bbd189d86dd915b3aec1b274","subject":"Remove non-inclusive language in pxe filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a1f8926cd161b770bbd189d86dd915b3aec1b274"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-09 18:07:18.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-15 14:45:28.000000000","tz":-420},"subject":"Respond so a ContentType is included","message":"Respond so a ContentType is included\n\nWhen sending a literal empty response, Flask does not include a\nContentType in the response. While in many cases, we don\u0027t need\nneed a ContentType nor expect one on the API client, Apache\nwebserver can treat this as an error and generate an Error\nindicating a Bad Gateway.\n\nHowever, when this Bad Gateway error occurs, the API/Conductor\nwere proceeding like there was no issue. The API client on the\nother hand thinks that a hard failure has occured.\n\nAlso adds some additional catches to provide additional logging\nwhich turned out not to be needed in this case, but it would be\nuseful for others.\n\nChange-Id: If2e7697e3fde58ab0a4193787e29d3acdca81ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/06457f154552a74ab2e373fa35be210b05803364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/06457f154552a74ab2e373fa35be210b05803364"}]},"branch":"refs/heads/master"},"c6cc4780f6f48f07cd1ff0d3144a430f66fb1470":{"kind":"REWORK","_number":5,"created":"2020-07-15 14:59:52.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/44/740344/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/44/740344/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/5"}}},"commit":{"parents":[{"commit":"a1f8926cd161b770bbd189d86dd915b3aec1b274","subject":"Remove non-inclusive language in pxe filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a1f8926cd161b770bbd189d86dd915b3aec1b274"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-09 18:07:18.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-15 14:59:48.000000000","tz":-420},"subject":"Respond so a ContentType is included","message":"Respond so a ContentType is included\n\nWhen sending a literal empty response, Flask does not include a\nContentType in the response. While in many cases, we don\u0027t need\nneed a ContentType nor expect one on the API client, Apache\nwebserver can treat this as an error and generate an Error\nindicating a Bad Gateway.\n\nHowever, when this Bad Gateway error occurs, the API/Conductor\nwere proceeding like there was no issue. The API client on the\nother hand thinks that a hard failure has occured.\n\nAlso adds some additional catches to provide additional logging\nwhich turned out not to be needed in this case, but it would be\nuseful for others.\n\nChange-Id: If2e7697e3fde58ab0a4193787e29d3acdca81ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/c6cc4780f6f48f07cd1ff0d3144a430f66fb1470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/c6cc4780f6f48f07cd1ff0d3144a430f66fb1470"}]},"branch":"refs/heads/master"},"f5cda854d2798e2350f574671e8faf788bd14afe":{"kind":"REWORK","_number":6,"created":"2020-07-15 16:22:00.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/44/740344/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/44/740344/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/6"}}},"commit":{"parents":[{"commit":"a1f8926cd161b770bbd189d86dd915b3aec1b274","subject":"Remove non-inclusive language in pxe filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a1f8926cd161b770bbd189d86dd915b3aec1b274"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-09 18:07:18.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-15 16:21:55.000000000","tz":-420},"subject":"Respond so a ContentType is included","message":"Respond so a ContentType is included\n\nWhen sending a literal empty response, Flask does not include a\nContentType in the response. While in many cases, we don\u0027t need\nneed a ContentType nor expect one on the API client, Apache\nwebserver can treat this as an error and generate an Error\nindicating a Bad Gateway.\n\nHowever, when this Bad Gateway error occurs, the API/Conductor\nwere proceeding like there was no issue. The API client on the\nother hand thinks that a hard failure has occured.\n\nAlso adds some additional catches to provide additional logging\nwhich turned out not to be needed in this case, but it would be\nuseful for others.\n\nChange-Id: If2e7697e3fde58ab0a4193787e29d3acdca81ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/f5cda854d2798e2350f574671e8faf788bd14afe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/f5cda854d2798e2350f574671e8faf788bd14afe"}]},"branch":"refs/heads/master"},"1a24facee6b4c20dfd8cfad76ff7b9cc7200f197":{"kind":"REWORK","_number":7,"created":"2020-07-16 16:46:15.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/44/740344/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/44/740344/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/7"}}},"commit":{"parents":[{"commit":"a1f8926cd161b770bbd189d86dd915b3aec1b274","subject":"Remove non-inclusive language in pxe filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a1f8926cd161b770bbd189d86dd915b3aec1b274"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-09 18:07:18.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-16 16:46:04.000000000","tz":-420},"subject":"DNM: Respond so a ContentType is included","message":"DNM: Respond so a ContentType is included\n\nWhen sending a literal empty response, Flask does not include a\nContentType in the response. While in many cases, we don\u0027t need\nneed a ContentType nor expect one on the API client, Apache\nwebserver can treat this as an error and generate an Error\nindicating a Bad Gateway.\n\nHowever, when this Bad Gateway error occurs, the API/Conductor\nwere proceeding like there was no issue. The API client on the\nother hand thinks that a hard failure has occured.\n\nAlso adds some additional catches to provide additional logging\nwhich turned out not to be needed in this case, but it would be\nuseful for others.\n\nChange-Id: If2e7697e3fde58ab0a4193787e29d3acdca81ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/1a24facee6b4c20dfd8cfad76ff7b9cc7200f197"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/1a24facee6b4c20dfd8cfad76ff7b9cc7200f197"}]},"branch":"refs/heads/master"},"deb4c145f21c1d008a834fccde1766df9c975bb8":{"kind":"REWORK","_number":8,"created":"2020-07-16 19:38:22.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/44/740344/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/44/740344/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/8"}}},"commit":{"parents":[{"commit":"a1f8926cd161b770bbd189d86dd915b3aec1b274","subject":"Remove non-inclusive language in pxe filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a1f8926cd161b770bbd189d86dd915b3aec1b274"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-09 18:07:18.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-16 19:38:17.000000000","tz":-420},"subject":"DNM: Respond so a ContentType is included","message":"DNM: Respond so a ContentType is included\n\nWhen sending a literal empty response, Flask does not include a\nContentType in the response. While in many cases, we don\u0027t need\nneed a ContentType nor expect one on the API client, Apache\nwebserver can treat this as an error and generate an Error\nindicating a Bad Gateway.\n\nHowever, when this Bad Gateway error occurs, the API/Conductor\nwere proceeding like there was no issue. The API client on the\nother hand thinks that a hard failure has occured.\n\nAlso adds some additional catches to provide additional logging\nwhich turned out not to be needed in this case, but it would be\nuseful for others.\n\nChange-Id: If2e7697e3fde58ab0a4193787e29d3acdca81ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/deb4c145f21c1d008a834fccde1766df9c975bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/deb4c145f21c1d008a834fccde1766df9c975bb8"}]},"branch":"refs/heads/master"},"7954be63b48d93c0b3e717d621c79cb26614de8e":{"kind":"REWORK","_number":9,"created":"2020-07-21 20:20:56.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/44/740344/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/44/740344/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/9"}}},"commit":{"parents":[{"commit":"a1f8926cd161b770bbd189d86dd915b3aec1b274","subject":"Remove non-inclusive language in pxe filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a1f8926cd161b770bbd189d86dd915b3aec1b274"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-09 18:07:18.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-21 20:20:41.000000000","tz":-420},"subject":"Respond so Apache HTTPd doesn\u0027t think the request failed","message":"Respond so Apache HTTPd doesn\u0027t think the request failed\n\nWhen sending a literal empty response, Flask does not include a\nContentType in the response. While in many cases, we don\u0027t need\nneed a ContentType nor expect one on the API client, Apache\nwebserver can treat this as an error and generate an Error\nindicating a Bad Gateway. When doing this, we also now include\nan empty JSON body in the response for 202 messages. For 204\nmessage errors, the message body is expected to be empty.\n\nHowever, when this Bad Gateway error occurs, the API/Conductor\nwere proceeding like there was no issue. The API client on the\nother hand thinks that a hard failure has occured.\n\nAlso adds some additional catches to provide additional logging\nwhich turned out not to be needed in this case, but it would be\nuseful for others.\n\nChange-Id: If2e7697e3fde58ab0a4193787e29d3acdca81ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/7954be63b48d93c0b3e717d621c79cb26614de8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/7954be63b48d93c0b3e717d621c79cb26614de8e"}]},"branch":"refs/heads/master"},"6e2cb60e778180a40418dbc2011fa25763c32363":{"kind":"REWORK","_number":10,"created":"2020-07-22 17:18:09.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/44/740344/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/44/740344/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/44/740344/10"}}},"commit":{"parents":[{"commit":"a1f8926cd161b770bbd189d86dd915b3aec1b274","subject":"Remove non-inclusive language in pxe filter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/a1f8926cd161b770bbd189d86dd915b3aec1b274"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-09 18:07:18.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-07-22 17:18:00.000000000","tz":-420},"subject":"Respond so Apache HTTPd doesn\u0027t think the request failed","message":"Respond so Apache HTTPd doesn\u0027t think the request failed\n\nWhen sending a literal empty response, Flask does not include a\nContentType in the response. While in many cases, we don\u0027t need\nneed a ContentType nor expect one on the API client, Apache\nwebserver can treat this as an error and generate an Error\nindicating a Bad Gateway. When doing this, we also now include\nan empty JSON body in the response for 202 messages. For 204\nmessage errors, the message body is expected to be empty.\n\nHowever, when this Bad Gateway error occurs, the API/Conductor\nwere proceeding like there was no issue. The API client on the\nother hand thinks that a hard failure has occured.\n\nAlso adds some additional catches to provide additional logging\nwhich turned out not to be needed in this case, but it would be\nuseful for others.\n\nChange-Id: If2e7697e3fde58ab0a4193787e29d3acdca81ebf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/6e2cb60e778180a40418dbc2011fa25763c32363"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/6e2cb60e778180a40418dbc2011fa25763c32363"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
