)]}'
{"id":"openstack%2Fironic-inspector~755613","triplet_id":"openstack%2Fironic-inspector~stable%2Fvictoria~Ib689686f3ea9e017e003a77d54f7ceba5a26c9e0","project":"openstack/ironic-inspector","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"Ib689686f3ea9e017e003a77d54f7ceba5a26c9e0","subject":"Update .gitreview for stable/victoria","status":"MERGED","created":"2020-10-01 18:42:18.000000000","updated":"2020-10-02 09:25:55.000000000","submitted":"2020-10-02 09:24:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755613-1601630677342-e81968aa","meta_rev_id":"337006f986da1f845ebbed20da1a073dace457f5","_number":755613,"virtual_id_number":755613,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-02 09:24:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-10-01 18:59:23.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-10-01 18:43:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":2,"date":"2020-10-02 08:04:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2020-10-02 08:04:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-01 18:43:06.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-10-01 18:59:23.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-02 08:04:50.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-10-02 09:24:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7312254ee6f9f4b85740cf739be825511788bcee","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-10-01 18:42:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9016e12a9778eed81385472425144a5960dabbe5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-10-01 18:43:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"941220e2833cfaa712aae3b8b3075172b443e736","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2020-10-01 18:59:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-8-ovb-1ctlr_1comp-featureset001 https://review.rdoproject.org/zuul/build/1879e29c834648ae8d9a6f2a467b11e0 : FAILURE in 15m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5b06cb30b3b21cb5bd939f6a98786b545b31aac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 20:27:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1987b90b79bb448a80450cc9927130b9 : SUCCESS in 6m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a48495292d5e413aa1bdb480a1f43c4d : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee6d111e55e240e0b746f6f304c9915a : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aebad54a9ade41d09b9f99d1f8df8508 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0114f681d7744c98b3f1281af9cf7613 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840b27accdff45ebbd01b35efd1c3c23 : SUCCESS in 7m 22s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/8603a744f96743178597b67659eb8c02 : SUCCESS in 1h 08m 24s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/d8a173f1ed4d4cfbb764338a6cd27942 : SUCCESS in 58m 19s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/8d6432bcebb4454aa82e54b6ea6907bf : SUCCESS in 43m 24s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/f348f5ff5f1b41769050b6ffeae509d9 : SUCCESS in 1h 06m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a2e9e9b09a53478d85ae7e6197da625a : SUCCESS in 5m 32s\n- bifrost-integration-tinyipa-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9b1638963234a8ebd9f16bc2cebd1b2 : SUCCESS in 22m 01s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/9d59e1b847d64462aa7f8c5a2062943a : SUCCESS in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02a16a21a9cb930998b1efc9079106a5df4bd067","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-10-02 08:04:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"06aa03e6dc354bb61ed889c54ebd432db2a6b28e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 08:05:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"325c0ba4105444c2729a8c73b733b35fe5ebba36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 09:24:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30ece4fcbd2e49cabb09d9d51c07430c : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21fc40fedb4a40beb9caec5a981cf10a : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2805477cb3c24c4d89a96d0110c10d2e : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b96d807b5144be5934cd8cbeac2bdcd : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24a0494b2875499c9aa9e426cc0b2a1f : SUCCESS in 8m 13s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/f3c3ba52032544bbaa05e96c48715f5d : SUCCESS in 1h 10m 49s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/36284c7778f7443298b387286659f342 : SUCCESS in 56m 10s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/a5ea282687684d129176548c4544412c : SUCCESS in 42m 33s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/e2f17bfab8544c5d80838e443595424b : SUCCESS in 57m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a72d69b294cf4b62b87ce68ccb13b863 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c88ddc89a62597afda2d6e91b57172369ff09203","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 09:24:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8d141b06e95c716bc92b928b98b32de76f4ed9d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-02 09:25:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ccdd084af2d441daf9ab68e5acba460 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f2f10e98f21eb9a39c5a246809b0d42ccca9f927","revisions":{"f2f10e98f21eb9a39c5a246809b0d42ccca9f927":{"kind":"REWORK","_number":1,"created":"2020-10-01 18:42:18.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/13/755613/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/13/755613/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/755613/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/755613/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/755613/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/13/755613/1"}}},"commit":{"parents":[{"commit":"e4273d1f0bbb152d9ab8b165eefc4b40f257afdf","subject":"Merge \"pxe filter - option to always block unknown hosts\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/e4273d1f0bbb152d9ab8b165eefc4b40f257afdf"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-01 18:42:16.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-01 18:42:16.000000000","tz":0},"subject":"Update .gitreview for stable/victoria","message":"Update .gitreview for stable/victoria\n\nChange-Id: Ib689686f3ea9e017e003a77d54f7ceba5a26c9e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/f2f10e98f21eb9a39c5a246809b0d42ccca9f927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/f2f10e98f21eb9a39c5a246809b0d42ccca9f927"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
