)]}'
{"id":"openstack%2Fironic-inspector~766950","triplet_id":"openstack%2Fironic-inspector~master~I352f21772ccd68b35806f46851f254edb7c24c60","project":"openstack/ironic-inspector","branch":"master","topic":"tox-light","hashtags":[],"change_id":"I352f21772ccd68b35806f46851f254edb7c24c60","subject":"Avoid a full install in tox environments that do not need it","status":"MERGED","created":"2020-12-14 13:19:44.000000000","updated":"2020-12-14 16:11:11.000000000","submitted":"2020-12-14 16:08:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"766950-tox-light","meta_rev_id":"0354e5cbae6e0a0448fe5093bbb9732e3bd3ae1a","_number":766950,"virtual_id_number":766950,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-14 16:08:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2020-12-14 13:40:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2020-12-14 14:42:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":1,"date":"2020-12-14 14:42:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-14 13:33:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-14 13:40:16.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-12-14 14:42:01.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"}],"messages":[{"id":"8443b2f41afe023d95275a4705d926350efec8c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-12-14 13:19:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f683faa25901d31b508407f72b41b058c46975e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 13:33:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1a2031f198234394843bd35394222f0c : SUCCESS in 3m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a61ba758d785491f88b5808f2e4a7cef : SUCCESS in 3m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f827fcf0a4d845ab9039ab5033861001 : SUCCESS in 5m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cfd7edf62a64c4eb720fc9ab12ba39d : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7754e62bb9644d68032a6b145f11a9f : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b19ff88fe6d549059ff857a13ff58335 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9474d7ca65e441e2afd2fdf74894bf13 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9907b72863084e20a3903e78779895cd : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a67e4d891994b8ca2791a4179e04d21 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4114a13713bd42688153aaee72da68e5 : SUCCESS in 4m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6ae010a19a27420c88869c4ba49ec467 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"030af712613c005be0f9d998c5f157c88366cc68","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-14 13:40:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7216d4e2cd8eef5411a746b1a93f04fecd60d02d","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2020-12-14 14:42:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"293927b136d6ff96b5306688f181d7c49b1cef03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 14:42:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"31c87e7195c7ba2cff47bdec6bbc8ca9ad72c7b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 16:08:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/00792e4aa9fa4e2096e77ad34e061c89 : SUCCESS in 3m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ed73bbe14bfb4cd397cddef38fec4c87 : SUCCESS in 2m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cac30232f20d445aa7e81895c3a54e54 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f9dd963dbd24e269c046444aae7bcdb : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97a8640888ae436ab76813f37bfb3bea : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddeafecc5c3b4cd087d05dc0fa3f7938 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6a905cc5a94e50b9b430490261cbcd : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17f9510545174078a47041c8d8908e71 : SUCCESS in 4m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e2c7f5f602c5457a9f1a728d0c22130d : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"dbbd37d57aec52754889bf3580cf590b7186b22d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 16:08:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0354e5cbae6e0a0448fe5093bbb9732e3bd3ae1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 16:11:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/2249ef15af2249409a124206f71eefe5 : SUCCESS in 1m 01s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c709a786d74d18ad5aecf4a42f034b : SUCCESS in 2m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9169fa5444c4661b69311a6ddf5f90e : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8d52a7789ac9eba2b5f8fa34e10922cd917f8e36","revisions":{"8d52a7789ac9eba2b5f8fa34e10922cd917f8e36":{"kind":"REWORK","_number":1,"created":"2020-12-14 13:19:44.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/50/766950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/50/766950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/50/766950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/50/766950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/50/766950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/50/766950/1"}}},"commit":{"parents":[{"commit":"8b8266b910d10c4aa1b1a74e7a00f8e92fac6089","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/8b8266b910d10c4aa1b1a74e7a00f8e92fac6089"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-12-14 13:18:46.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2020-12-14 13:18:46.000000000","tz":60},"subject":"Avoid a full install in tox environments that do not need it","message":"Avoid a full install in tox environments that do not need it\n\nUsage of usedevelop\u003dTrue results in all requirements.txt dependencies\nalways installed (without constraints) even when requirements.txt\nis not explicitly added. Override usedevelop for these environments.\n\nAdd requirements.txt to the \u0027docs\u0027 environment since it imports\nthe whole ironic in the process.\n\nCreate a separate doc/requirements.txt.\n\nChange-Id: I352f21772ccd68b35806f46851f254edb7c24c60\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/8d52a7789ac9eba2b5f8fa34e10922cd917f8e36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/8d52a7789ac9eba2b5f8fa34e10922cd917f8e36"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
