)]}'
{"id":"openstack%2Fironic-inspector~776213","triplet_id":"openstack%2Fironic-inspector~master~I83c420d0e91e00f583a28833f4f710cf70b01fa8","project":"openstack/ironic-inspector","branch":"master","topic":"fix-logging-id","hashtags":[],"change_id":"I83c420d0e91e00f583a28833f4f710cf70b01fa8","subject":"Fix node id vs uuid in processing_logger_prefix","status":"MERGED","created":"2021-02-17 15:14:25.000000000","updated":"2021-02-18 15:00:09.000000000","submitted":"2021-02-18 14:48:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"776213-fix-logging-id","meta_rev_id":"9683dfad845db0359e089de69a87a26575ed7287","_number":776213,"virtual_id_number":776213,"owner":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10288,"name":"Aleksey","email":"alexz.kh.ua@gmail.com","username":"alexz"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-18 14:47:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":10288,"name":"Aleksey","email":"alexz.kh.ua@gmail.com","username":"alexz"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":2,"date":"2021-02-17 18:40:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-18 13:26:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":10288,"name":"Aleksey","email":"alexz.kh.ua@gmail.com","username":"alexz"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-18 13:26:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10288,"name":"Aleksey","email":"alexz.kh.ua@gmail.com","username":"alexz"},{"value":0,"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10288,"name":"Aleksey","email":"alexz.kh.ua@gmail.com","username":"alexz"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 15:18:24.000000000","updated_by":{"_account_id":10288,"name":"Aleksey","email":"alexz.kh.ua@gmail.com","username":"alexz"},"reviewer":{"_account_id":10288,"name":"Aleksey","email":"alexz.kh.ua@gmail.com","username":"alexz"},"state":"REVIEWER"},{"updated":"2021-02-17 15:18:31.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-02-17 16:09:07.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-02-17 16:39:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cbdc7440fdaae6b7b4e78aef9cb993b089b1b3a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-02-17 15:14:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b460791b7fe481e364bf6183644502745ceea034","author":{"_account_id":10288,"name":"Aleksey","email":"alexz.kh.ua@gmail.com","username":"alexz"},"date":"2021-02-17 15:18:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f3f1589f820f07cce70bea395a59044b7697b849","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-02-17 15:18:31.000000000","message":"Patch Set 1: Code-Review+2\n\nI wonder if it\u0027s possible to unit test this fix.","accounts_in_message":[],"_revision_number":1},{"id":"3a8dfb54ce5de6edce0557a248857c7ea35c57fa","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-02-17 16:01:54.000000000","message":"Patch Set 1: Workflow-1\n\n@Dmitry, I\u0027ll have to modify the tests anyway, currently they are failing\n\nIN unit tests, the node object is mocked out, and thus whatever we set on it - works, the usual bane of mocks and unit tests.. :-/\n\nMay be we can instantiate a real Node object from openstacksdk but mock out some of its parts that access outside...","accounts_in_message":[],"_revision_number":1},{"id":"3a0bf683757b06400ee6c7924edcb2ac7a545525","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 16:39:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ce7ec185e0f48fea9941af895ff0ddc : FAILURE in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e4206e476248b68d844e0e4adf0046 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/536802680c384612bb0c9fda19cd861f : FAILURE in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cfed0db2a8b747eb82cf304f664f3bd8 : FAILURE in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbf1b142150346ed8fd476adeb16895c : FAILURE in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edf1736745104133b0217e2e421992ec : SUCCESS in 11m 34s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/f5d02e7bfdba4ec2b76236811e390c72 : SUCCESS in 1h 18m 55s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/4b28fa39359646ed994e874fc467b67a : SUCCESS in 58m 05s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/78e3d1aace3a447db4ec84a8e40c4e84 : SUCCESS in 47m 25s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/8b8ecee5d8a145ffada2cbd35c71b0c3 : SUCCESS in 1h 02m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c3d2652b8dbe4536a50d7cef00efac4b : SUCCESS in 6m 21s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b39427f60c44b59a21ac0843fdadc52 : SUCCESS in 34m 00s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/203c0086ddb04882835416bded0189ed : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf0d9445d03f177370f00661201f93247bc93dd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-02-17 16:42:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"761ca450f5dae590e87c6ad077f3e949dff5e794","author":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"date":"2021-02-17 16:45:22.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nfound solution for unit tests - use spec_set [0] instead of simply spec\n\nhttps://docs.python.org/3.6/library/unittest.mock.html#unittest.mock.Mock\n\nwe should be using it more...","accounts_in_message":[],"_revision_number":2},{"id":"f38c2528b91ce974aec3c8d3c523f96fefae5790","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 18:08:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7067f7fef6454f09802a508ac921d713 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0603200a5aaa4c52a245ecf160534c86 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/919a174a68dd4434a40a4c0bc73495af : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9abfab47c424043af6ee4fcabbc6035 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84a8f85b74fd411ba8d7a5700d101c33 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e00280afb59421493a0647cb1993862 : SUCCESS in 7m 44s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/a4ee74be6aba4382a71080e8828199ea : SUCCESS in 1h 05m 35s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/093e96c218004166ace4a345e0b465b5 : SUCCESS in 50m 49s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/3c9ee2ec052f4b3ebebb9a3bca0b1291 : SUCCESS in 40m 15s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/090284e49fd349a39b57adccc51c8340 : SUCCESS in 1h 04m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/705319158ca643c9ac67fa530217c8ec : SUCCESS in 6m 00s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e18f10735924fb2acc9e5248e0f09f5 : SUCCESS in 25m 01s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/ff0ba0328d824a93b1c6201f3a7326d3 : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"51fc80884f8f554928bf6628fb6ffc64ed39c2ef","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-02-17 18:40:33.000000000","message":"Patch Set 2: Code-Review+2\n\nthanks!","accounts_in_message":[],"_revision_number":2},{"id":"a7cebdbc6dd7c1b7ff5457a6c563105aa1041ae0","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-02-18 13:26:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1f530702025be4d5617381f37769fc7c50d61658","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 13:29:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"048746058836ae1109d49eaaf1600399a8dd5679","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 14:47:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/998cb27cf68c45fca696e3d52f6281cc : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/462ab162a2664bd1845b4175ce8f23df : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a295eadb7ebf41e9b8d68d8e27e78bb0 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c63559203d194a7d8e168fa8db51408a : SUCCESS in 8m 32s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/107fb379dabc4c0f8a6251ecfe5e64b2 : SUCCESS in 1h 08m 06s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/2194b17c1a0a499c98aa236b70f67752 : SUCCESS in 51m 48s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/356e095a662649a7ae8da4817a3fcb54 : SUCCESS in 45m 20s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/27fa11e844da48dc8c100eab6adb7cbf : SUCCESS in 55m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/175c5126b3d842b888242998cd8fdeb2 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":2},{"id":"7f3f36f23e8cb76860b8936739aca01d13d4d28d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 14:48:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9683dfad845db0359e089de69a87a26575ed7287","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 15:00:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b16de81ce0f34ba8927887c02e93ed6a : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5c79d7552a76adc27b7246b411312ab832a0f28b","revisions":{"90158a5d6474ddee841b0a40ca1f1c250fecaf80":{"kind":"REWORK","_number":1,"created":"2021-02-17 15:14:25.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/13/776213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/13/776213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/776213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/776213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/776213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/13/776213/1"}}},"commit":{"parents":[{"commit":"8d06567a6e233d616c3c6cd1118c993b1481ff86","subject":"Merge \"Add upgrade check, and json2yaml policy handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/8d06567a6e233d616c3c6cd1118c993b1481ff86"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-02-17 15:12:20.000000000","tz":120},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-02-17 15:12:20.000000000","tz":120},"subject":"Fix node id vs uuid in processing_logger_prefix","message":"Fix node id vs uuid in processing_logger_prefix\n\nthis is a follow-up to Icd3de82877c6a53d32b4c9fd3e500d3cd9d7fb17\n\none more place was identified as trying to use uuid attr of a Node\nobject from openstacksdk instead of id attr.\n\nChange-Id: I83c420d0e91e00f583a28833f4f710cf70b01fa8\nStory: 2008379\nTask: 41300\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/90158a5d6474ddee841b0a40ca1f1c250fecaf80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/90158a5d6474ddee841b0a40ca1f1c250fecaf80"}]},"branch":"refs/heads/master"},"5c79d7552a76adc27b7246b411312ab832a0f28b":{"kind":"REWORK","_number":2,"created":"2021-02-17 16:42:55.000000000","uploader":{"_account_id":9542,"name":"Pavlo Shchelokovskyy","email":"pshchelokovskyy@mirantis.com","username":"pshchelo"},"ref":"refs/changes/13/776213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/13/776213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/776213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/776213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/13/776213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/13/776213/2"}}},"commit":{"parents":[{"commit":"8d06567a6e233d616c3c6cd1118c993b1481ff86","subject":"Merge \"Add upgrade check, and json2yaml policy handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/8d06567a6e233d616c3c6cd1118c993b1481ff86"}]}],"author":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-02-17 15:12:20.000000000","tz":120},"committer":{"name":"Pavlo Shchelokovskyy","email":"shchelokovskyy@gmail.com","date":"2021-02-17 16:40:58.000000000","tz":120},"subject":"Fix node id vs uuid in processing_logger_prefix","message":"Fix node id vs uuid in processing_logger_prefix\n\nthis is a follow-up to Icd3de82877c6a53d32b4c9fd3e500d3cd9d7fb17\n\none more place was identified as trying to use uuid attr of a Node\nobject from openstacksdk instead of id attr.\n\nIn tests, use stricter mocking with spec_set (that actually fails to\naccess or set an atribute of a mock that is absent from the spec class\nor instance) to guard against future possible changes.\n\nChange-Id: I83c420d0e91e00f583a28833f4f710cf70b01fa8\nStory: 2008379\nTask: 41300\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/5c79d7552a76adc27b7246b411312ab832a0f28b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/5c79d7552a76adc27b7246b411312ab832a0f28b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
