)]}'
{"id":"openstack%2Fironic-inspector~896325","triplet_id":"openstack%2Fironic-inspector~stable%2Fyoga~I0d194091be22e8401d379b7ffa720f8004cca3d0","project":"openstack/ironic-inspector","branch":"stable/yoga","topic":"bug/2036455-stable/2023.1-stable/zed-stable/yoga","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2024-02-06 14:58:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0d194091be22e8401d379b7ffa720f8004cca3d0","subject":"Handle bracketed IPv6 redfish_address","status":"ABANDONED","created":"2023-09-24 10:02:38.000000000","updated":"2024-02-06 14:58:08.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":896324,"cherry_pick_of_patch_set":1,"meta_rev_id":"e26dffdb76a8a7cd29d4f3b1338dd176b017b448","_number":896325,"virtual_id_number":896325,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-09-24 11:20:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2023-09-27 13:42:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-24 11:20:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-27 13:42:01.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2024-02-06 14:58:08.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"6da2f283dcdf48c731b03d7cfab0e0ddf552ff39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2023-09-24 10:02:38.000000000","message":"Patch Set 1: Cherry Picked from branch stable/zed.","accounts_in_message":[],"_revision_number":1},{"id":"352049c3b48d96cd201352ab4b80d3f5bfdc3b41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-24 11:20:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/958f02e9739b4cca8ba1caadf25dd5e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecd560a21e38421d8cf255d37ed78e27 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08ce2d7343164bfab737e49e902ef182 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a346710e84a46fd94a5a91e4a0703d0 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ecd76bf8406454192ec9ac88f599b3f : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3477fcd244a4a928826b59534b02745 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12459e325bed43b7bbf079199910a946 : SUCCESS in 5m 27s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/c31a08b6e9ed4c0aa7e71031dc182365 : SUCCESS in 49m 33s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/1cc29cab60c74fa8a25ee5528024dfa3 : SUCCESS in 41m 48s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/df7fcf2c441b48f39f3dfe466fae2ca3 : SUCCESS in 51m 13s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/102cec67b94d498faf9dd5759987b857 : FAILURE in 1h 12m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af9e0f9fe6324d5f973d830ded653734 : SUCCESS in 4m 54s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b45769deec844269a94503d365b1996a : SUCCESS in 20m 44s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/218124c282ec4678acb92703a44d105d : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"334ee1dbf48aea053dcfbadcb74c9598fb492a42","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2023-09-25 09:44:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"663755ddc0b43042cfb77f35dca21c2246474036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 11:05:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a05d87eb98644008dffb5aabbafd0a4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77cb62fa5c82409babaa8be0a087e865 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abe1b0259bfe41cab8dfbf5c338514ac : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76f7b963a3ec49d19404e937d925cb0f : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40329e6ae4794036b2a43d341984cf1b : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce7398c2ffc544078686c3e9b1ca1d5e : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e70e2198888447d933ab0efd83d209c : SUCCESS in 4m 26s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/ea6f12de7f9c40ae9fa3ac164bf81ef6 : SUCCESS in 58m 09s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/e3b86f4e63194fa7896fafe1582a77c9 : SUCCESS in 54m 18s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/60883a545f3847d0ae50df6f4e79c3d4 : SUCCESS in 45m 51s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/2702a58615f74e248fcbfba4e5145cf7 : FAILURE in 1h 13m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f768a0d53623495aa9ae107eb1f5fe71 : SUCCESS in 5m 19s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2db407187617426cbeea9312589febf4 : SUCCESS in 24m 22s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/a92fbe5cf3e44f35bd3415a5970ca6bd : SUCCESS in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"556a9c7c0fb06c2136f6e207b5f74abfe3453f5f","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-27 13:42:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f66bb24bc53e357f81db4400f50941bcc1881437","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2023-10-12 07:54:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5fbed4c3ed0f554b4b088623ce79e790d9c8eda7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 09:17:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b2857bd6572144c8b93e779d1d98427a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b032ffa700bb4e908e05d5521aa6d979 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fee2af90e9e4134ab48933a32b7ada2 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdeeaf9225f6434e918594ef5d99b83f : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f009c465b8347a6875d6a9616578c73 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a18b80a73ab4f5ca67aca6e31535185 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33d71d5d2474406b9966287f84d79e63 : SUCCESS in 3m 51s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/c1b647a3f89f4e74acd49b9cf12c4b33 : FAILURE in 25m 49s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/aa2f4c99d6bc4dc8980f6a99840bda80 : SUCCESS in 1h 04m 32s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/f9ee594207554e8e8eb1c447fb21a8a4 : SUCCESS in 53m 36s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/aa5d5556be6843198617382f59dda290 : FAILURE in 1h 15m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5e5ad4929ac643e9959cea761eda6d66 : SUCCESS in 5m 09s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b999b18d24d149eb9073e4b836a98292 : SUCCESS in 29m 10s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/cab1e44860304408acff67bf263c7973 : SUCCESS in 3m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dbf42a0637a6e073f45eb46e215831d7e4a51667","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-11-15 22:25:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8a3e82d1324093e0242950d2d46115021d264eff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 23:55:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6255b5805a2c43c790a24d637267a9d0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9855a7e864a48e1af62bbb137fb01ae : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3006aeb07424c289f15ad52fcbc8f2c : SUCCESS in 3m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09f386d85c454ca98bfb98ff086aea0d : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/919c656f9b844abcaba5a63a4765d765 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a79ec765ea9490a90414fcf0b04b0bc : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c562b546ab114fdfad0ea11588074a8d : SUCCESS in 4m 29s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/b521d32cc29b43749c2df04b18005cb9 : SUCCESS in 1h 23m 38s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/317ed11c1c0c4b41bc85faccd8843e92 : SUCCESS in 45m 20s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/0c02ae92e8e748eaac3505ec6261c66c : SUCCESS in 38m 20s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/df30353387d0462a950d840b68d39a19 : FAILURE in 54m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/15e5944dc8c74ebdb1f5f7f3c71d6a88 : SUCCESS in 6m 10s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc5f1b4d6c1e4f94bda8ba5cc244a90a : SUCCESS in 21m 54s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/28d19867e6f94c049a08e88e75983bde : SUCCESS in 3m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b888836d1a8ca92d9357cf09cbf1c83a64943a7","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2023-11-16 08:49:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54776c9e57c3d4b32f0d1000f33b9083651030ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 10:07:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a437326011d48a4b00591cfea12e540\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09a18d74418045128f1f146f987298d8 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd6510b61974f9e8ffcd9ed8cef13ce : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4028e9d629c4730a34f41ba0f8c39e3 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61ab3e9baff4444582e114a91f0dd4b8 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b48d0d8d73544af4a501a38c5df8e656 : SUCCESS in 7m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16151c7121b14587b9ba049cbde972c5 : SUCCESS in 4m 41s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/2a3141bf7b074b6a9bba070dbb7442fd : SUCCESS in 1h 17m 22s\n- ironic-inspector-tempest https://zuul.opendev.org/t/openstack/build/da06dad86227441baf5a20ad70ec660e : SUCCESS in 51m 46s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/6065279bff574a44a1c171d1f326989c : SUCCESS in 52m 17s\n- ironic-inspector-tempest-managed-non-standalone https://zuul.opendev.org/t/openstack/build/6375faf42c1c41a6b18d7850230d0965 : FAILURE in 1h 06m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3c184d696b1c4c9784c4c126b9f6de16 : SUCCESS in 5m 57s\n- bifrost-integration-tinyipa-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d1abe7cc9d04dd1a8d982f7d88935cc : SUCCESS in 29m 07s (non-voting)\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/25625fcee6774e6b8b00bd8570314b06 : SUCCESS in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e26dffdb76a8a7cd29d4f3b1338dd176b017b448","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-02-06 14:58:08.000000000","message":"Abandoned\n\nstable/yoga branch of openstack/ironic-inspector is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/yoga if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c03f81ac7d2e1316a9b8d0cdac793362b6610c8a","revisions":{"c03f81ac7d2e1316a9b8d0cdac793362b6610c8a":{"kind":"REWORK","_number":1,"created":"2023-09-24 10:02:38.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/25/896325/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/25/896325/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/25/896325/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/25/896325/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/25/896325/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/25/896325/1"}}},"commit":{"parents":[{"commit":"561fed27c7959b72aa74a46026a62fc1f941139b","subject":"Make reno ignore bugfix eol tags","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/561fed27c7959b72aa74a46026a62fc1f941139b"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2023-09-18 18:19:03.000000000","tz":120},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2023-09-24 10:02:38.000000000","tz":0},"subject":"Handle bracketed IPv6 redfish_address","message":"Handle bracketed IPv6 redfish_address\n\nIf redfish_address is in brackets, unwrap it\nand check that it is a valid IPv6 address.\nIf that is the case use the unwrapped address\nto avoid \"Name or service not known\".\n\nCloses-Bug: #2036455\nChange-Id: I0d194091be22e8401d379b7ffa720f8004cca3d0\n(cherry picked from commit 278ba52bf99a0f0c60224e518c6d952c2ac93698)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/c03f81ac7d2e1316a9b8d0cdac793362b6610c8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/c03f81ac7d2e1316a9b8d0cdac793362b6610c8a"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
