)]}'
{"id":"openstack%2Fironic-inspector~930802","triplet_id":"openstack%2Fironic-inspector~master~I012ea6a97e6b837d58fb10ea9161eeab2805cc66","project":"openstack/ironic-inspector","branch":"master","hashtags":[],"change_id":"I012ea6a97e6b837d58fb10ea9161eeab2805cc66","subject":"Fix invalid assertion","status":"ABANDONED","created":"2024-09-28 06:31:40.000000000","updated":"2024-09-28 11:49:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9e1c3cbca953b54e810dc2a323e53cc4b683fd77","_number":930802,"virtual_id_number":930802,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-28 07:31:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-28 07:31:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b33d3dfb5aea5d266de4c38c65d62c3f66cc17d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-28 06:31:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9a809efefa6fea7979509ae2f52e82679a11f1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-28 07:31:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bc0fa17aca945d3b17bcbdd3989f834\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/937a587fba2442738d89c6ec79390377 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76a5d530358746f4b2a85b3dd2ac3f29 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27fc07d5ef7741208cc3e687ba1f3214 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/946368305b284459bc72a3d5c79c0f70 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/572784e6da5f44f58fac4e8f8f9aa7ec : SUCCESS in 7m 06s\n- ironic-inspector-grenade https://zuul.opendev.org/t/openstack/build/72eef3136f1742ef9e11b29ef50e8e35 : SUCCESS in 59m 06s\n- ironic-inspector-tempest-discovery https://zuul.opendev.org/t/openstack/build/335414f8c1f543acaba55e5c97ce0851 : SUCCESS in 42m 12s\n- openstack-tox-functional https://zuul.opendev.org/t/openstack/build/cfff0440f7624e0b9ec39e4d1f36f238 : SUCCESS in 4m 02s\n- ironic-inspector-tox-bandit https://zuul.opendev.org/t/openstack/build/be659b26f03148a4891b47d5311ef434 : SUCCESS in 3m 45s (non-voting)\n- ironic-inspector-tox-codespell https://zuul.opendev.org/t/openstack/build/8eafd7bd14ab4f8493b5cc89924b6732 : FAILURE in 3m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e1c3cbca953b54e810dc2a323e53cc4b683fd77","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-28 11:49:57.000000000","message":"Abandoned\n\nthis is duplicate of https://review.opendev.org/c/openstack/ironic-inspector/+/929815","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ae7062a17c322a91f4deb4b40d210cfcbfe3b81b","revisions":{"ae7062a17c322a91f4deb4b40d210cfcbfe3b81b":{"kind":"REWORK","_number":1,"created":"2024-09-28 06:31:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/02/930802/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-inspector","ref":"refs/changes/02/930802/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/02/930802/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/02/930802/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-inspector refs/changes/02/930802/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-inspector refs/changes/02/930802/1"}}},"commit":{"parents":[{"commit":"61672f074438d9dd291ad485ab36eff8e095ee60","subject":"Merge \"Fix versions in release notes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/61672f074438d9dd291ad485ab36eff8e095ee60"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 06:24:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-28 06:31:27.000000000","tz":540},"subject":"Fix invalid assertion","message":"Fix invalid assertion\n\n... to resolve the following error now detected in Python 3.12 job.\n\nAttributeError: \u0027called_once_with\u0027 is not a valid assertion. Use a spec\nfor the mock if \u0027called_once_with\u0027 is meant to be an attribute..\nDid you mean: \u0027assert_called_once_with\u0027?\n\nChange-Id: I012ea6a97e6b837d58fb10ea9161eeab2805cc66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/ae7062a17c322a91f4deb4b40d210cfcbfe3b81b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-inspector/commit/ae7062a17c322a91f4deb4b40d210cfcbfe3b81b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
