)]}'
{"id":"openstack%2Fironic-lib~663957","triplet_id":"openstack%2Fironic-lib~master~I887af3efde729a33e05eff779012175889649033","project":"openstack/ironic-lib","branch":"master","topic":"mdns-interfaces","hashtags":[],"change_id":"I887af3efde729a33e05eff779012175889649033","subject":"mdns: allow specifying which interfaces to listen on","status":"MERGED","created":"2019-06-07 14:26:13.000000000","updated":"2019-06-12 07:33:34.000000000","submitted":"2019-06-12 07:33:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663957-1560324814222-7fce86ad","meta_rev_id":"d998987c320a0bc0a1f88a650e064bf6162bf6d2","_number":663957,"virtual_id_number":663957,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"date":"2019-06-08 10:41:18.000000000","_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":2,"date":"2019-06-12 07:33:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"date":"2019-06-10 10:00:00.000000000","_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2019-06-10 04:57:03.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-06-10 06:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-10 10:33:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2019-06-12 05:53:45.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-08 10:41:18.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-06-10 06:28:33.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"},{"updated":"2019-06-10 10:33:18.000000000","updated_by":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"reviewer":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"state":"REVIEWER"},{"updated":"2019-06-12 05:53:45.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2019-06-12 07:33:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc83f436fd2a0d941f71d17c94c6636752c74605","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-07 14:26:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3212106800d9e995f049e1b988ff669f48107971","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-07 15:58:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover http://logs.openstack.org/57/663957/1/check/openstack-tox-cover/3d1282d/cover/ : SUCCESS in 5m 12s\n- openstack-tox-lower-constraints http://logs.openstack.org/57/663957/1/check/openstack-tox-lower-constraints/822b2bc/ : SUCCESS in 5m 43s\n- openstack-tox-pep8 http://logs.openstack.org/57/663957/1/check/openstack-tox-pep8/e569958/ : SUCCESS in 7m 55s\n- openstack-tox-py27 http://logs.openstack.org/57/663957/1/check/openstack-tox-py27/58d5489/ : SUCCESS in 4m 14s\n- openstack-tox-py36 http://logs.openstack.org/57/663957/1/check/openstack-tox-py36/e84b3ee/ : SUCCESS in 5m 07s\n- openstack-tox-py37 http://logs.openstack.org/57/663957/1/check/openstack-tox-py37/31ea6f3/ : SUCCESS in 5m 22s\n- openstack-tox-docs http://logs.openstack.org/57/663957/1/check/openstack-tox-docs/04fea30/html/ : SUCCESS in 4m 00s\n- ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src-python2 http://logs.openstack.org/57/663957/1/check/ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src-python2/fc47229/ : SUCCESS in 1h 01m 36s\n- ironic-lib-partition-bios-ipmi-direct-tinyipa-src-python2 http://logs.openstack.org/57/663957/1/check/ironic-lib-partition-bios-ipmi-direct-tinyipa-src-python2/76b3b7b/ : SUCCESS in 1h 03m 02s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2 http://logs.openstack.org/57/663957/1/check/ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2/386fecb/ : SUCCESS in 1h 03m 34s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src http://logs.openstack.org/57/663957/1/check/ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src/d4735ca/ : SUCCESS in 56m 37s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src http://logs.openstack.org/57/663957/1/check/ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src/9c8e8d3/ : SUCCESS in 1h 06m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d020b6b7f23dfd5b190d500b1a5cc3bf05bfe8c4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-06-08 10:41:18.000000000","message":"Patch Set 1:\n\nI know we don\u0027t really use release notes for ironic-lib, but I have the very strong feeling we should do it, but this is probably a discussion for the meeting.\n\nAnyway, should we link this to the mdns story in storyboard?\nhttps://storyboard.openstack.org/#!/story/2005393","accounts_in_message":[],"_revision_number":1},{"id":"38468926490e8eec1e4c0b77db19cbacb624d5a6","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-06-10 04:57:03.000000000","message":"Patch Set 1: Code-Review+2\n\nThank you Dmitry.","accounts_in_message":[],"_revision_number":1},{"id":"02866f7df4a0142588ef7c9d8a3eb38874b583ce","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-06-10 06:28:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\nIf the ironic-lib is considered a deliverable, I think we\u0027ll need release notes, but it seems we never thought so. There was only one release note added in https://review.opendev.org/#/c/648606/, however we even don\u0027t have the job for generating and publishing release notes.","accounts_in_message":[],"_revision_number":1},{"id":"27dc039aa4d52f87d23b7719362d3b52a49eb89e","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-10 10:00:00.000000000","message":"Patch Set 1:\n\nironic-lib is an internal deliverable, that\u0027s why we haven\u0027t considered adding release notes for it.","accounts_in_message":[],"_revision_number":1},{"id":"d9ac9c22bb77600c790ee4f2b619a0f43a6b5fca","author":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"date":"2019-06-10 10:33:18.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a52b2b63997f1096345621a0ad191e2bf9330610","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2019-06-12 05:53:45.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1a6f9ba2c975d837506b43df8bc2d955cae6216b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 05:54:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85ca888a2bb892c6e3d145d1aabd0bb29e500f13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 07:33:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints http://logs.openstack.org/57/663957/1/gate/openstack-tox-lower-constraints/30ed94a/ : SUCCESS in 7m 30s\n- openstack-tox-pep8 http://logs.openstack.org/57/663957/1/gate/openstack-tox-pep8/05db498/ : SUCCESS in 4m 39s\n- openstack-tox-py27 http://logs.openstack.org/57/663957/1/gate/openstack-tox-py27/9967507/ : SUCCESS in 7m 24s\n- openstack-tox-py36 http://logs.openstack.org/57/663957/1/gate/openstack-tox-py36/221f57c/ : SUCCESS in 4m 29s\n- openstack-tox-py37 http://logs.openstack.org/57/663957/1/gate/openstack-tox-py37/0fc9f1b/ : SUCCESS in 5m 03s\n- openstack-tox-docs http://logs.openstack.org/57/663957/1/gate/openstack-tox-docs/ca25cd6/html/ : SUCCESS in 4m 49s\n- ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src-python2 http://logs.openstack.org/57/663957/1/gate/ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src-python2/20d2c19/ : SUCCESS in 59m 40s\n- ironic-lib-partition-bios-ipmi-direct-tinyipa-src-python2 http://logs.openstack.org/57/663957/1/gate/ironic-lib-partition-bios-ipmi-direct-tinyipa-src-python2/202ea84/ : SUCCESS in 59m 44s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2 http://logs.openstack.org/57/663957/1/gate/ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2/67ce595/ : SUCCESS in 1h 04m 08s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src http://logs.openstack.org/57/663957/1/gate/ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src/1563c8f/ : SUCCESS in 1h 01m 35s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src http://logs.openstack.org/57/663957/1/gate/ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src/5a8bf89/ : SUCCESS in 56m 26s","accounts_in_message":[],"_revision_number":1},{"id":"38d13c02ea398e52e0dc231e276d386414903267","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 07:33:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56887c9660b05a6b9eec7f956398d304c8f0d7ff","revisions":{"56887c9660b05a6b9eec7f956398d304c8f0d7ff":{"kind":"REWORK","_number":1,"created":"2019-06-07 14:26:13.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/57/663957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/57/663957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/57/663957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/57/663957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/57/663957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/57/663957/1"}}},"commit":{"parents":[{"commit":"617056b2a36d039179271cb24fd6157ee2505ef5","subject":"mdns: allow quoting in the [mdns]params configuration option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/617056b2a36d039179271cb24fd6157ee2505ef5"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-07 14:25:42.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-06-07 14:25:42.000000000","tz":120},"subject":"mdns: allow specifying which interfaces to listen on","message":"mdns: allow specifying which interfaces to listen on\n\nChange-Id: I887af3efde729a33e05eff779012175889649033\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/56887c9660b05a6b9eec7f956398d304c8f0d7ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/56887c9660b05a6b9eec7f956398d304c8f0d7ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
