)]}'
{"id":"openstack%2Fironic-lib~665054","triplet_id":"openstack%2Fironic-lib~master~I6ed1ec6c6b0e833fddc68e2931be053c8880fd65","project":"openstack/ironic-lib","branch":"master","topic":"python3-train","hashtags":[],"change_id":"I6ed1ec6c6b0e833fddc68e2931be053c8880fd65","subject":"Update Python 3 test runtimes for Train","status":"MERGED","created":"2019-06-13 03:38:52.000000000","updated":"2019-06-13 09:04:19.000000000","submitted":"2019-06-13 09:04:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665054-1560416659245-22ee9161","meta_rev_id":"e96fb4cc3ab71fc13cd7d58840a62965f897677f","_number":665054,"virtual_id_number":665054,"owner":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2019-06-13 09:04:19.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2019-06-13 07:07:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-13 07:31:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":2,"date":"2019-06-13 07:40:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":1,"date":"2019-06-13 07:40:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-13 07:07:06.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-06-13 07:31:33.000000000","updated_by":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"reviewer":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"state":"REVIEWER"},{"updated":"2019-06-13 07:40:15.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2019-06-13 09:04:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1700dae6a59773c727c705a137010e39b4a28644","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2019-06-13 03:38:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da6fc39884f1cb3cb08e3a415c76d2b5fb6628b0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 05:00:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack http://logs.openstack.org/54/665054/1/check/test-release-openstack/5c2b1e6/ : SUCCESS in 3m 30s\n- openstack-tox-cover http://logs.openstack.org/54/665054/1/check/openstack-tox-cover/0c8a197/cover/ : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/665054/1/check/openstack-tox-lower-constraints/6abe039/ : SUCCESS in 5m 48s\n- openstack-tox-pep8 http://logs.openstack.org/54/665054/1/check/openstack-tox-pep8/8d5f922/ : SUCCESS in 4m 15s\n- openstack-tox-py27 http://logs.openstack.org/54/665054/1/check/openstack-tox-py27/09ed2c8/ : SUCCESS in 4m 27s\n- openstack-tox-py36 http://logs.openstack.org/54/665054/1/check/openstack-tox-py36/b50cf96/ : SUCCESS in 4m 46s\n- openstack-tox-py37 http://logs.openstack.org/54/665054/1/check/openstack-tox-py37/3ab36bf/ : SUCCESS in 4m 40s\n- openstack-tox-docs http://logs.openstack.org/54/665054/1/check/openstack-tox-docs/ccf078c/html/ : SUCCESS in 5m 19s\n- ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src-python2 http://logs.openstack.org/54/665054/1/check/ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src-python2/4b14b06/ : SUCCESS in 1h 02m 44s\n- ironic-lib-partition-bios-ipmi-direct-tinyipa-src-python2 http://logs.openstack.org/54/665054/1/check/ironic-lib-partition-bios-ipmi-direct-tinyipa-src-python2/d1d606e/ : SUCCESS in 1h 07m 01s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2 http://logs.openstack.org/54/665054/1/check/ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2/784cced/ : SUCCESS in 1h 01m 21s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src http://logs.openstack.org/54/665054/1/check/ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src/b53c216/ : SUCCESS in 1h 04m 59s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src http://logs.openstack.org/54/665054/1/check/ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src/35a8af9/ : SUCCESS in 57m 39s","accounts_in_message":[],"_revision_number":1},{"id":"fc9cc588bcb1cc394abf603019d6bd49b85c96e7","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-06-13 07:07:06.000000000","message":"Patch Set 1: Code-Review+2\n\nthanks Kaifeng","accounts_in_message":[],"_revision_number":1},{"id":"ca1fb58b5bd0f80036904ef225071ff125205e83","author":{"_account_id":10206,"name":"Madhuri Kumari","email":"madhuri.kumari@intel.com","username":"Madhuri"},"date":"2019-06-13 07:31:33.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"e7643e5681c8df38b4b8786d44e299c4fc2938dd","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-06-13 07:40:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7ee6d7b3ef0d08d1fb8d273e8f3062a3964acbd5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 07:40:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37584918729cd2384f4ad8c1f90e1a25cc8e7f93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:04:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aa7f9cde3959bac53f78ba5e9ee8c5ae87caadaa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-13 09:04:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack http://logs.openstack.org/54/665054/1/gate/test-release-openstack/7fc0168/ : SUCCESS in 3m 34s\n- openstack-tox-lower-constraints http://logs.openstack.org/54/665054/1/gate/openstack-tox-lower-constraints/b4fadca/ : SUCCESS in 6m 11s\n- openstack-tox-pep8 http://logs.openstack.org/54/665054/1/gate/openstack-tox-pep8/1a9e6ba/ : SUCCESS in 4m 05s\n- openstack-tox-py27 http://logs.openstack.org/54/665054/1/gate/openstack-tox-py27/77ec626/ : SUCCESS in 5m 06s\n- openstack-tox-py36 http://logs.openstack.org/54/665054/1/gate/openstack-tox-py36/7ad36c1/ : SUCCESS in 5m 50s\n- openstack-tox-py37 http://logs.openstack.org/54/665054/1/gate/openstack-tox-py37/747d7c4/ : SUCCESS in 4m 51s\n- openstack-tox-docs http://logs.openstack.org/54/665054/1/gate/openstack-tox-docs/01c70b5/html/ : SUCCESS in 4m 43s\n- ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src-python2 http://logs.openstack.org/54/665054/1/gate/ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src-python2/9cf9cbe/ : SUCCESS in 58m 18s\n- ironic-lib-partition-bios-ipmi-direct-tinyipa-src-python2 http://logs.openstack.org/54/665054/1/gate/ironic-lib-partition-bios-ipmi-direct-tinyipa-src-python2/2c74b94/ : SUCCESS in 59m 39s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2 http://logs.openstack.org/54/665054/1/gate/ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2/d6decff/ : SUCCESS in 1h 02m 23s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src http://logs.openstack.org/54/665054/1/gate/ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src/70a4def/ : SUCCESS in 1h 02m 10s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src http://logs.openstack.org/54/665054/1/gate/ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src/a606105/ : SUCCESS in 1h 02m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04d9ebf605c7788381c10eb9d4fbe6fd6f9a63b8","revisions":{"04d9ebf605c7788381c10eb9d4fbe6fd6f9a63b8":{"kind":"REWORK","_number":1,"created":"2019-06-13 03:38:52.000000000","uploader":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"ref":"refs/changes/54/665054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/54/665054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/665054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/665054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/665054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/54/665054/1"}}},"commit":{"parents":[{"commit":"56887c9660b05a6b9eec7f956398d304c8f0d7ff","subject":"mdns: allow specifying which interfaces to listen on","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/56887c9660b05a6b9eec7f956398d304c8f0d7ff"}]}],"author":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-06-13 03:32:50.000000000","tz":480},"committer":{"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","date":"2019-06-13 03:32:50.000000000","tz":480},"subject":"Update Python 3 test runtimes for Train","message":"Update Python 3 test runtimes for Train\n\nOne of train goals to have consistent python3 jobs.\n\nhttps://governance.openstack.org/tc/goals/train/python3-updates.html\n\nChange-Id: I6ed1ec6c6b0e833fddc68e2931be053c8880fd65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/04d9ebf605c7788381c10eb9d4fbe6fd6f9a63b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/04d9ebf605c7788381c10eb9d4fbe6fd6f9a63b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
