)]}'
{"id":"openstack%2Fironic-lib~673822","triplet_id":"openstack%2Fironic-lib~stable%2Frocky~Ic3963e5c4a1886aeabed6c5994599e72040a251e","project":"openstack/ironic-lib","branch":"stable/rocky","topic":"p-scheme-stable/rocky","hashtags":[],"change_id":"Ic3963e5c4a1886aeabed6c5994599e72040a251e","subject":"Use last digit to determine paritition naming scheme","status":"MERGED","created":"2019-07-31 13:57:45.000000000","updated":"2019-08-09 17:48:10.000000000","submitted":"2019-08-09 17:48:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673822-1565372890018-92a0a85e","meta_rev_id":"a7d73073a1ec5d8bbc82fc475d1009166042fa68","_number":673822,"virtual_id_number":673822,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":2,"date":"2019-08-09 17:48:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-08-09 10:56:37.000000000","_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"date":"2019-07-31 13:57:45.000000000","_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-09 15:50:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-31 13:57:45.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-08-09 10:56:37.000000000","updated_by":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"reviewer":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"state":"REVIEWER"},{"updated":"2019-08-09 17:48:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6a92768e396a3d5af1111019c76b5b63096968d","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-07-31 13:57:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0b6dca69fbe7387bb0489f2e8dd67becfa212f75","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-31 15:18:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/22/673822/1/check/openstack-tox-pep8/123c45c/ : SUCCESS in 4m 21s\n- openstack-tox-py27 http://logs.openstack.org/22/673822/1/check/openstack-tox-py27/d188965/ : SUCCESS in 4m 54s\n- openstack-tox-py35 http://logs.openstack.org/22/673822/1/check/openstack-tox-py35/534487e/ : SUCCESS in 4m 03s\n- build-openstack-sphinx-docs http://logs.openstack.org/22/673822/1/check/build-openstack-sphinx-docs/855063c/html/ : SUCCESS in 4m 37s\n- ironic-lib-tempest-partition-agent_ipmitool http://logs.openstack.org/22/673822/1/check/ironic-lib-tempest-partition-agent_ipmitool/8da3c56/ : SUCCESS in 1h 06m 21s\n- ironic-lib-tempest-partition-pxe_ipmitool http://logs.openstack.org/22/673822/1/check/ironic-lib-tempest-partition-pxe_ipmitool/b4ee47e/ : SUCCESS in 53m 58s\n- ironic-lib-tempest-wholedisk-agent_ipmitool http://logs.openstack.org/22/673822/1/check/ironic-lib-tempest-wholedisk-agent_ipmitool/972bdff/ : SUCCESS in 1h 02m 17s\n- ironic-lib-tempest-wholedisk-pxe_ipmitool http://logs.openstack.org/22/673822/1/check/ironic-lib-tempest-wholedisk-pxe_ipmitool/481b778/ : SUCCESS in 1h 00m 57s\n- openstack-tox-lower-constraints http://logs.openstack.org/22/673822/1/check/openstack-tox-lower-constraints/e0cb0bf/ : SUCCESS in 6m 03s\n- openstack-tox-cover http://logs.openstack.org/22/673822/1/check/openstack-tox-cover/622f53e/cover/ : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0c5ba8669a856aeb31801ef334583ab6b53b88f0","author":{"_account_id":10343,"name":"Jim Rollenhagen","email":"jim@jimrollenhagen.com","username":"jimrollenhagen"},"date":"2019-08-09 10:56:37.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good, waiting for more recent branches to land.","accounts_in_message":[],"_revision_number":1},{"id":"ad5769835a6025fce2e78a6252972ee018ab4608","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2019-08-09 15:50:07.000000000","message":"Patch Set 1: Workflow+1\n\nstein merged","accounts_in_message":[],"_revision_number":1},{"id":"07d228b7b16f2118230eed9e1f73adaa2589a50f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 15:50:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d79d312f20e9b49b8aa24d9f1682b452c108b839","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 17:48:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://logs.opendev.org/22/673822/1/gate/openstack-tox-pep8/8dd4b46/ : SUCCESS in 3m 31s\n- openstack-tox-py27 https://logs.opendev.org/22/673822/1/gate/openstack-tox-py27/87bb5ad/ : SUCCESS in 5m 05s\n- openstack-tox-py35 https://logs.opendev.org/22/673822/1/gate/openstack-tox-py35/329fd22/ : SUCCESS in 4m 54s\n- build-openstack-sphinx-docs https://logs.opendev.org/22/673822/1/gate/build-openstack-sphinx-docs/02c8237/html/ : SUCCESS in 4m 29s\n- ironic-lib-tempest-partition-agent_ipmitool https://logs.opendev.org/22/673822/1/gate/ironic-lib-tempest-partition-agent_ipmitool/f316647/ : SUCCESS in 1h 19m 18s\n- ironic-lib-tempest-partition-pxe_ipmitool https://logs.opendev.org/22/673822/1/gate/ironic-lib-tempest-partition-pxe_ipmitool/1b246eb/ : SUCCESS in 53m 12s\n- ironic-lib-tempest-wholedisk-agent_ipmitool https://logs.opendev.org/22/673822/1/gate/ironic-lib-tempest-wholedisk-agent_ipmitool/4fb06c7/ : SUCCESS in 56m 34s\n- ironic-lib-tempest-wholedisk-pxe_ipmitool https://logs.opendev.org/22/673822/1/gate/ironic-lib-tempest-wholedisk-pxe_ipmitool/cf459ea/ : SUCCESS in 59m 02s\n- openstack-tox-lower-constraints https://logs.opendev.org/22/673822/1/gate/openstack-tox-lower-constraints/bd743e0/ : SUCCESS in 6m 05s","accounts_in_message":[],"_revision_number":1},{"id":"44f5e46841881609c42cabc3a10a2f032c4c9ae5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 17:48:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ca0cf67c141f5db1b7333b2d0ae07de04261ada","revisions":{"3ca0cf67c141f5db1b7333b2d0ae07de04261ada":{"kind":"REWORK","_number":1,"created":"2019-07-31 13:57:45.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/22/673822/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/22/673822/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/22/673822/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/22/673822/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/22/673822/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/22/673822/1"}}},"commit":{"parents":[{"commit":"9731ae35bb9d449b434fc82293b778683b33643b","subject":"Include partiton name and flags from parted output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/9731ae35bb9d449b434fc82293b778683b33643b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-07-04 11:31:42.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2019-07-31 13:57:45.000000000","tz":0},"subject":"Use last digit to determine paritition naming scheme","message":"Use last digit to determine paritition naming scheme\n\nThis matches the behaviour of the linux kernel, see:\n\nhttps://github.com/torvalds/linux/blob/d3511f53bb2475f2a4e8460bee5a1ae6dea2a433/block/partition-generic.c#L39\n\nChange-Id: Ic3963e5c4a1886aeabed6c5994599e72040a251e\nStory: 2006154\nTask: 35656\n(cherry picked from commit cd51319528bb2333139c7cefb68e7e6e27df740a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/3ca0cf67c141f5db1b7333b2d0ae07de04261ada"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/3ca0cf67c141f5db1b7333b2d0ae07de04261ada"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
