)]}'
{"id":"openstack%2Fironic-lib~684771","triplet_id":"openstack%2Fironic-lib~master~I2241bfd11c66fb84b8840ab76563af52fe93e5ad","project":"openstack/ironic-lib","branch":"master","hashtags":[],"change_id":"I2241bfd11c66fb84b8840ab76563af52fe93e5ad","subject":"Move ironic-lib.filters to etc/ironic/rootwrap.d","status":"MERGED","created":"2019-09-25 15:01:28.000000000","updated":"2019-10-16 13:11:22.000000000","submitted":"2019-09-27 09:04:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684771-1569575082746-fe29816b","meta_rev_id":"15e3a4506a4149a64a522941e4e64dedc4a2c614","_number":684771,"virtual_id_number":684771,"owner":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 09:04:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"all":[{"value":2,"date":"2019-09-26 13:59:43.000000000","_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":2,"date":"2019-09-27 07:44:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":1,"date":"2019-09-27 07:44:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-26 13:59:43.000000000","updated_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"reviewer":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"state":"REVIEWER"},{"updated":"2019-09-27 07:44:03.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2019-09-27 09:04:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9b5974239e2c530704a7ac5287c1e9ad5c35022a","author":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"date":"2019-09-25 15:01:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"29e79169cdacda9a18a83eba8c70ef93c813a8b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 16:22:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd8b943b4ef64032a81e831f0855a5a6 : SUCCESS in 3m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95587f2211734eac8f9c26bcf3dca409 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4ef0a98f3b740049480cd8c5d57999e : SUCCESS in 2m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4a4f16f7e5354caf88e8b1b85a4f556a : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d41b8a1955c44caa85f019d9ea31a35 : SUCCESS in 4m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/854f78fdc630423b9fca09b691693f52 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebbaf69e3d3a46b49eb8ed9186bcb37f : SUCCESS in 4m 43s\n- ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/f0470043ad134a45947fbb9ecf31c7fb : SUCCESS in 1h 13m 14s\n- ironic-lib-partition-bios-ipmi-direct-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/798c9291592a4d9086a81e3f989dc3df : SUCCESS in 1h 02m 19s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/26f68dd0dd4e4d3babfd166d90075096 : SUCCESS in 1h 05m 38s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/e42b1dd77dba4cc192f337f656a23f05 : SUCCESS in 1h 12m 49s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/c8fb80290f4c4ba0b3aa907960be227b : SUCCESS in 58m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1f095c284de7342c2c891579219041ee339993bb","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2019-09-26 13:59:43.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks. I\u0027m wondering if anything is assuming the original location, but hoping/guessing not.","accounts_in_message":[],"_revision_number":1},{"id":"3caa181895054f7ed1a5d1fe51f84aeca85365f2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-09-27 07:44:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nthis type of files are by default \"per project\", their locations depend on rootwrap.conf, the correct path is under /etc/ironic/rootwrap.d/\nif anything breaks because of this change, then the path in the configuration must be changed to the one here","accounts_in_message":[],"_revision_number":1},{"id":"33802bb1108c2648fb622267c99169e43c1f89f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 07:44:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0d9055cee523a5070bf38ef9fbd5d3fc74c1d5e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 09:04:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4977116194445239914181468ca2a25 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2776bc23dac045fea20f9560440fc712 : SUCCESS in 3m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c4edd0e6d7c242348ff0d21273c9e8d6 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2ec1d2f3ccf40f582079761477b6465 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7fc5d55fa2874fb4a104992cc2c0a8a5 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb97e3736cb74bfeb063b360cfc56fc1 : SUCCESS in 4m 14s\n- ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/8b6b0e18546b4a7887970af44530b23a : SUCCESS in 1h 09m 07s\n- ironic-lib-partition-bios-ipmi-direct-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/b0e0fce307ea46e382ad8be38509a921 : SUCCESS in 1h 09m 36s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/0f9779653edc4a86bdc36cb8f97f265e : SUCCESS in 59m 02s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/009b7a790510420e88e1958e288064fe : SUCCESS in 1h 14m 32s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/ebe0acc16c8b4788bfeb256bfaed5860 : SUCCESS in 1h 07m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ec7a6fd1fe645ec1d6a46a43f6239aa93830657c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 09:04:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"375c0b2473e2c42b7607f1b880f71dbe4a8cc3ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 09:07:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cdb2ebe4a2c4c17baa66c970c5941a6 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1},{"id":"bb7929569d6824dc490bb7e57dc35f7be2c6f22d","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2019-09-30 15:47:26.000000000","message":"Patch Set 1:\n\n@Riccardo. The path is configurable [1] (or supposed to be). Whether anyone configures it, I don\u0027t know.\n\n[1] https://opendev.org/openstack/ironic/src/commit/0d1a929f9e94fc453ce81bec5b2804d4cfb557d0/ironic/conf/default.py#L333","accounts_in_message":[],"_revision_number":1},{"id":"f36c23f14b5ed4b4e1c7da275424d44244828680","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-01 07:14:35.000000000","message":"Patch Set 1:\n\n\u003e @Riccardo. The path is configurable [1] (or supposed to be).\n \u003e Whether anyone configures it, I don\u0027t know.\n \u003e \n \u003e [1] https://opendev.org/openstack/ironic/src/commit/0d1a929f9e94fc453ce81bec5b2804d4cfb557d0/ironic/conf/default.py#L333\n\nHey Ruby, this is a different thing.\nWhat\u0027s configurable there is the rootwrap configuration file path, while this patch is about the filters file path.\nI don\u0027t think the latter is configurable.","accounts_in_message":[],"_revision_number":1},{"id":"0f9d09c9a185de88b82f0c3249f0d04524bd978c","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2019-10-16 13:08:26.000000000","message":"Patch Set 1:\n\n@Riccardo. The path to the rootwrap.conf file is configurable. And the rootwrap.conf file itself, includes a config that specifies the filter path: https://opendev.org/openstack/ironic/src/commit/a2ae57c457962e6fa5e590c0de19f4d457637881/etc/ironic/rootwrap.conf#L7","accounts_in_message":[],"_revision_number":1},{"id":"15e3a4506a4149a64a522941e4e64dedc4a2c614","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2019-10-16 13:11:22.000000000","message":"Patch Set 1:\n\n\u003e @Riccardo. The path to the rootwrap.conf file is configurable. And\n \u003e the rootwrap.conf file itself, includes a config that specifies the\n \u003e filter path: https://opendev.org/openstack/ironic/src/commit/a2ae57c457962e6fa5e590c0de19f4d457637881/etc/ironic/rootwrap.conf#L7\n\nYes, of course it\u0027s configurable inside rootwrap.conf, I meant just not directly configurable by any option in ironic :)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4d1162f19b2db4d3c89a5ee4acd828644d5ee8e","revisions":{"c4d1162f19b2db4d3c89a5ee4acd828644d5ee8e":{"kind":"REWORK","_number":1,"created":"2019-09-25 15:01:28.000000000","uploader":{"_account_id":7102,"name":"Thomas Bechtold","email":"thomas.bechtold@chainguard.dev","username":"toabctl"},"ref":"refs/changes/71/684771/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/71/684771/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/71/684771/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/71/684771/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/71/684771/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/71/684771/1"}}},"commit":{"parents":[{"commit":"0c8dbfaf269c413bb0dbdfd1563fa5766a429a69","subject":"Merge \"Build pdf doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/0c8dbfaf269c413bb0dbdfd1563fa5766a429a69"}]}],"author":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-09-25 15:00:02.000000000","tz":120},"committer":{"name":"Thomas Bechtold","email":"tbechtold@suse.com","date":"2019-09-25 15:01:26.000000000","tz":120},"subject":"Move ironic-lib.filters to etc/ironic/rootwrap.d","message":"Move ironic-lib.filters to etc/ironic/rootwrap.d\n\nThis is the path where ironic searches for rootwrap filter files by\ndefault.\nThis is important because ironic itself is going to remove [1] its\ncopy of the ironic-lib.filters file.\n\n[1] https://review.opendev.org/#/c/684270\n\nChange-Id: I2241bfd11c66fb84b8840ab76563af52fe93e5ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/c4d1162f19b2db4d3c89a5ee4acd828644d5ee8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/c4d1162f19b2db4d3c89a5ee4acd828644d5ee8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
