)]}'
{"id":"openstack%2Fironic-lib~716998","triplet_id":"openstack%2Fironic-lib~master~Ie9fe810b639d60caf03d972eb966b2ab8401501a","project":"openstack/ironic-lib","branch":"master","topic":"ironic-convert-dib","hashtags":[],"change_id":"Ie9fe810b639d60caf03d972eb966b2ab8401501a","subject":"Convert jobs to dib","status":"MERGED","created":"2020-04-02 13:21:02.000000000","updated":"2020-04-08 08:43:32.000000000","submitted":"2020-04-08 08:36:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716998-1586334990921-9e742fff","meta_rev_id":"79df8f97a3b6d4871937f2c3f659a6fa7bb447f4","_number":716998,"virtual_id_number":716998,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-08 08:36:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2020-04-08 06:34:09.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-07 11:15:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":2,"date":"2020-04-07 11:06:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2020-04-08 06:34:09.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-07 11:06:14.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-04-07 11:15:45.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-04-08 06:34:09.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-04-08 08:36:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"433f1af3bd7c8e671be9642f1df7d8c6240fdd0e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-02 13:21:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e29e190769484864b4f15b65bbcf120814d7f08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 13:26:11.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/ironic-lib on branch master.  The error was:\n\n  Job ironic-python-agent-build-image-dib not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - openstack-cover-jobs\n        - openstack-lower-constraints-jobs\n        - openstack-python3-ussuri-jobs\n        - publish-openstack-docs-pti\n      check:\n        jobs:\n          - ironic-lib-wholedisk-bios-ipmi-direct-dib-src\n          - ironic-lib-partition-bios-ipmi-direct-dib-src\n          - ironic-lib-partition-bios-ipmi-iscsi-dib-src\n          - ironic-lib-wholedisk-bios-ipmi-iscsi-dib-src\n      gate:\n        queue: ironic\n        jobs:\n          - ironic-lib-wholedisk-bios-ipmi-direct-dib-src\n          - ironic-lib-partition-bios-ipmi-direct-dib-src\n          - ironic-lib-partition-bios-ipmi-iscsi-dib-src\n          - ironic-lib-wholedisk-bios-ipmi-iscsi-dib-src\n      post:\n        jobs:\n          - ironic-python-agent-build-image-dib\n\n  in \"openstack/ironic-lib/zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"92b1425bc0744d391e3dbd712c930fce8ec8d724","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-02 14:00:05.000000000","message":"Patch Set 1:\n\nprobably shouldn\u0027t have changed that one before changing it in ipa","accounts_in_message":[],"_revision_number":1},{"id":"35c8c84074e1eeafb3029f0ba367b911cea648fe","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-02 14:00:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3d48141ce86a9539d224cb565279c0ff43917d6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 15:20:06.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/757705d193d14c0e9ab972237198f05f : SUCCESS in 7m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3a554a877bee4cd6916273891f9bf300 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2ab1d7ca75456f8eaa2ec149847c13 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b33ffba0561949229a9023038cf91592 : SUCCESS in 7m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5ddcf05ab63c42a693da912ab3e0df03 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9866d1668fec41d1b778146bbef94c9d : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab85bd8e60cb4ba0b77b6e00035d3963 : SUCCESS in 10m 24s\n- ironic-lib-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/055534dc5d924aaaa09b5181acebc38e : FAILURE in 1h 05m 31s\n- ironic-lib-partition-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/63dc820a3e8a4aadb3a5473862b89b1b : FAILURE in 1h 09m 38s\n- ironic-lib-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/3c396f455e1048509432a75eb2ca8b0c : FAILURE in 1h 08m 13s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/b560f7d0d8394c47a6b326659b7bd7bd : FAILURE in 1h 01m 01s","accounts_in_message":[],"_revision_number":2},{"id":"48cc6af8d10309247c2cd6c4d8930622436f4f7b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-02 15:58:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"70d4f0ab4afd7b930907cf957154d73473966450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 18:56:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75a9a8c6d8134cf186b42df367daccc0 : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52de2cfe2b4b4e8c9adeb0fd0cd0a308 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71868807aeb2409f9b870107c914e5c1 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43e289c940f34addb9a267ac07a23b9c : SUCCESS in 5m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fff2dc0533d3447ea84b0ffa14dee19a : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f6ec0fb74ad4831a250093ba0c2b507 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5739a7bd6fc4af68fb5bd9010029120 : SUCCESS in 6m 44s\n- ironic-lib-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/0c55e2f36f6b4c7db31b9117d657bdf1 : SUCCESS in 1h 21m 00s\n- ironic-lib-partition-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/7bd8835af2ce4f1aafbc2ca94c482629 : FAILURE in 1h 01m 55s\n- ironic-lib-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/7adc347b4d0d4085b0b0c8c9f7d03792 : SUCCESS in 1h 08m 57s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/c429eaf26540442b98bdeca3cdc574dd : SUCCESS in 1h 07m 08s","accounts_in_message":[],"_revision_number":3},{"id":"fb7787a03660536b06604419b142ae3637815c61","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-03 12:50:29.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"26a6f34e3e1889ff763b3627e87a91cd813190e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-03 14:09:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db6bb8e4d2b84a8daea333794ec2e3e6 : SUCCESS in 4m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e57940942c7f4da296ee9371f8ee164c : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/278ac2de115c4986b64b72bb6e24a9aa : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13ae23b217d64c0381b07cfb3fa4f789 : SUCCESS in 5m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24a738ac0e6341bfaa665725ab3391d3 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fcd4d17873e49d8bbfa5b1cca650d85 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c49417a0ff584824a5c8bd87c7d727f9 : SUCCESS in 8m 47s\n- ironic-lib-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/def70a19b0574540b74d7abd3f3477cb : SUCCESS in 1h 16m 54s\n- ironic-lib-partition-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/76b358e4c2744478a43306c1febad3ea : FAILURE in 1h 05m 27s\n- ironic-lib-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/bb740172fcaa49159ea563983f957efa : SUCCESS in 1h 00m 31s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/b72ef0341d84434c81985a877303f57a : SUCCESS in 1h 04m 53s","accounts_in_message":[],"_revision_number":4},{"id":"6bc491f8107645a60e65c507dbc12d4b8fa71255","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-06 12:44:06.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"61ca1ee58677be4328cf29f80512e58d5d5fe447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-06 13:59:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/827a75ea84da404ab3da14ff04eb51d8 : SUCCESS in 3m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ef6c1fc1773b474eb917677656103b0c : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3f4413c4cd84ca4b1bfd783a498f195 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1897c3a322843d6a5e62b43116e3f82 : SUCCESS in 6m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/596820d1106f4a1e83072a2ca93e90c3 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8682b8ff93ee48c8b40e7a6d90fad012 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102f8ef8c9a048b88f0e5cdd73e60e27 : SUCCESS in 6m 12s\n- ironic-lib-wholedisk-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/a7314d0ba25d4c50b1bcddef599f6a3e : SUCCESS in 1h 07m 29s\n- ironic-lib-partition-bios-ipmi-direct-dib-src https://zuul.opendev.org/t/openstack/build/19029d03f87a4b848147c1d6bed03c7e : SUCCESS in 1h 09m 22s\n- ironic-lib-partition-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/95339d3811044ad9a02c0fd98651f5cd : SUCCESS in 1h 07m 51s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-dib-src https://zuul.opendev.org/t/openstack/build/7ddb7a7cc3d345629024333d06068f43 : SUCCESS in 1h 02m 16s","accounts_in_message":[],"_revision_number":4},{"id":"b1faeaa87d00c31695f90003e0de23c449b2f668","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-06 14:08:26.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5ca8d90ab3e243c1e7a8a96282464bfadce87446","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-06 16:07:18.000000000","message":"Patch Set 4: Code-Review+1\n\nLet\u0027s remove \"dib\" from names maybe?","accounts_in_message":[],"_revision_number":4},{"id":"12bf077def1c4e5f5dc427e0d2c63badc049643f","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-04-07 07:34:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"60504da95cf28bdfb5411f88fa7263a96324c43c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-07 08:55:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/283ee331c721476388d40443a289b00f : SUCCESS in 4m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60d168963b3e48edae8eff910712b116 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48a6f3d99d6e46119058c7ece9c6f197 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/896beb4b36e1456eaaa7e446709d70f6 : SUCCESS in 4m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0bb9328cba024f06ac9114acddc1300c : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3d4a994081c4fa7b821cd5cebbdba1d : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d208612016834eaf88a43aaf94d52456 : SUCCESS in 6m 43s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/0be078c4af8146b4b4d065d8487dd8cb : SUCCESS in 1h 12m 20s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/7141b86e1188438caff35610d6c2e6e7 : SUCCESS in 1h 14m 25s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/334cda04280b41f6819e297d05f6d1ef : SUCCESS in 1h 03m 59s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/d6ac451b64ba4482ba245459edec1258 : SUCCESS in 1h 11m 23s","accounts_in_message":[],"_revision_number":5},{"id":"8a174a33da4e49330c9c3970c72805da3bc2b7b8","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-04-07 11:06:14.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"8bc827053d96b0faa8e2455026e54d4cb77d8b9b","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-04-07 11:15:45.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"e95c31a02dfada48e9054dfc2582acc040c7af8a","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-04-08 06:34:09.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThank you Riccardo","accounts_in_message":[],"_revision_number":5},{"id":"f80c9d2500924d37aa9707436efc60aebe0b5884","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 06:34:28.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"bd1284a8e8404c23afe41ea32bfea373f86de2fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 08:36:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"0105234f4d9993bcf3356bf80ff4e3165086663e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 08:36:30.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0a8f5bfe0aa47c2ab181f51107268ec : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e68edf663a1a4476b8aa1eeb202fe3ec : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0945ecd86e214764a0c9a10a9ea08029 : SUCCESS in 3m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d16a8b9f1f3749a1b45997e0e2d01805 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfa6e3c342454bf99b7d85be3794ee8e : SUCCESS in 6m 20s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/18e267dfa38442c4a9c9b0d8eee95911 : SUCCESS in 1h 03m 31s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/fc0cbd44714e4417b608c56c174d82ee : SUCCESS in 1h 09m 52s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/45d5fd2bfdfe40cfa4beb7b80795a8c4 : SUCCESS in 1h 09m 30s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/95cf7453cbe34a099bd2f5c4df409130 : SUCCESS in 1h 10m 07s","accounts_in_message":[],"_revision_number":5},{"id":"db1488f3245fdb9ab5489632b8645ff54e864d35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-08 08:43:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120a97808ab846eaabda4e55043e6125 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"0838b15f61efd967eeadd2a32b4c2367f18a46be","revisions":{"28c8dbb2256045c5e24ae43cf7740523a448db79":{"kind":"REWORK","_number":1,"created":"2020-04-02 13:21:02.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/98/716998/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/98/716998/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/1"}}},"commit":{"parents":[{"commit":"f6cdde207ad0d2c80339640bc9c58a6ef1856e51","subject":"[trivial] add description of ignored flake8 checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/f6cdde207ad0d2c80339640bc9c58a6ef1856e51"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-02 13:20:09.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-02 13:20:51.000000000","tz":120},"subject":"Convert jobs to dib","message":"Convert jobs to dib\n\nChange-Id: Ie9fe810b639d60caf03d972eb966b2ab8401501a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/28c8dbb2256045c5e24ae43cf7740523a448db79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/28c8dbb2256045c5e24ae43cf7740523a448db79"}]},"branch":"refs/heads/master"},"3d5d8b924a84c61ad88f0620f91caf095437bd97":{"kind":"REWORK","_number":2,"created":"2020-04-02 14:00:55.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/98/716998/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/98/716998/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/2"}}},"commit":{"parents":[{"commit":"f6cdde207ad0d2c80339640bc9c58a6ef1856e51","subject":"[trivial] add description of ignored flake8 checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/f6cdde207ad0d2c80339640bc9c58a6ef1856e51"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-02 13:20:09.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-02 14:00:48.000000000","tz":120},"subject":"Convert jobs to dib","message":"Convert jobs to dib\n\nChange-Id: Ie9fe810b639d60caf03d972eb966b2ab8401501a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/3d5d8b924a84c61ad88f0620f91caf095437bd97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/3d5d8b924a84c61ad88f0620f91caf095437bd97"}]},"branch":"refs/heads/master"},"bdfe5e9b079e3e3f497b20d3c75ae71f7dceaa34":{"kind":"REWORK","_number":3,"created":"2020-04-02 15:58:19.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/98/716998/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/98/716998/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/3"}}},"commit":{"parents":[{"commit":"f6cdde207ad0d2c80339640bc9c58a6ef1856e51","subject":"[trivial] add description of ignored flake8 checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/f6cdde207ad0d2c80339640bc9c58a6ef1856e51"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-02 13:20:09.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-02 15:58:11.000000000","tz":120},"subject":"Convert jobs to dib","message":"Convert jobs to dib\n\nChange-Id: Ie9fe810b639d60caf03d972eb966b2ab8401501a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/bdfe5e9b079e3e3f497b20d3c75ae71f7dceaa34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/bdfe5e9b079e3e3f497b20d3c75ae71f7dceaa34"}]},"branch":"refs/heads/master"},"6e8241e0e62fb22bcde81562e987de38eabc646c":{"kind":"REWORK","_number":4,"created":"2020-04-03 12:50:29.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/98/716998/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/98/716998/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/4"}}},"commit":{"parents":[{"commit":"f6cdde207ad0d2c80339640bc9c58a6ef1856e51","subject":"[trivial] add description of ignored flake8 checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/f6cdde207ad0d2c80339640bc9c58a6ef1856e51"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-02 13:20:09.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-03 12:50:19.000000000","tz":120},"subject":"Convert jobs to dib","message":"Convert jobs to dib\n\nChange-Id: Ie9fe810b639d60caf03d972eb966b2ab8401501a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/6e8241e0e62fb22bcde81562e987de38eabc646c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/6e8241e0e62fb22bcde81562e987de38eabc646c"}]},"branch":"refs/heads/master"},"0838b15f61efd967eeadd2a32b4c2367f18a46be":{"kind":"REWORK","_number":5,"created":"2020-04-07 07:34:25.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/98/716998/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/98/716998/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/98/716998/5"}}},"commit":{"parents":[{"commit":"f6cdde207ad0d2c80339640bc9c58a6ef1856e51","subject":"[trivial] add description of ignored flake8 checks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/f6cdde207ad0d2c80339640bc9c58a6ef1856e51"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-02 13:20:09.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-04-07 07:34:14.000000000","tz":120},"subject":"Convert jobs to dib","message":"Convert jobs to dib\n\nChange-Id: Ie9fe810b639d60caf03d972eb966b2ab8401501a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/0838b15f61efd967eeadd2a32b4c2367f18a46be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/0838b15f61efd967eeadd2a32b4c2367f18a46be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
