)]}'
{"id":"openstack%2Fironic-lib~748246","triplet_id":"openstack%2Fironic-lib~master~I3a686b00cb2a6cb7cde4a26001d6d9016aab08eb","project":"openstack/ironic-lib","branch":"master","hashtags":[],"change_id":"I3a686b00cb2a6cb7cde4a26001d6d9016aab08eb","subject":"Remove the unused coding style modules","status":"MERGED","created":"2020-08-26 14:34:24.000000000","updated":"2020-08-28 07:12:29.000000000","submitted":"2020-08-28 07:10:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748246-1598598606259-8a9bf4a0","meta_rev_id":"9ef945ef58b0b605b20a9613240ef9097a200fcf","_number":748246,"virtual_id_number":748246,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-28 07:10:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":2,"date":"2020-08-28 05:17:15.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":1,"date":"2020-08-26 23:22:00.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-27 15:50:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":1,"date":"2020-08-28 05:17:15.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-26 23:22:00.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2020-08-27 15:50:39.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-08-28 05:17:15.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-08-28 07:10:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"73c0cc118d8719820a5ae2184331348a3aed9707","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-08-26 14:34:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98fb93f3cace0a6d4ade4587612f715744a5a928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-26 16:09:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/cabd35d1d9b048a497067043af5d44e4 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba228669834f449480aa0ab65aea4f3e : SUCCESS in 5m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/74555c15a098495589e47e601db64fc3 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/817db7b10bcb4a00a09e97579b312ccf : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a550c760f331465ca24ef4cb02eec1f4 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/415c582a537d46ce92f763fdacf86f7e : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d53fa4c438a468e925183d9c52dfc34 : SUCCESS in 6m 17s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/7d77b8ec2a02477fa5cc60257465784a : SUCCESS in 1h 25m 58s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/fc98c7a0e5084121b1617cd7c850cc2b : SUCCESS in 1h 19m 33s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/d18f184163974ad48146264aca7933ee : SUCCESS in 1h 26m 47s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/b938dc9cd8fe4999b9b2793f0b184e81 : SUCCESS in 1h 27m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2819ae9b08575d7cf1898bcfd47d7b9609f968fa","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-08-26 23:22:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0defec1032ae33b90715c6fac0cc0afb26f8e670","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-08-27 15:50:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"be84f1a44bacb9ae1fdb9e829289fb9b6fe61b7f","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-08-28 05:17:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThank you.","accounts_in_message":[],"_revision_number":1},{"id":"6239e1c6c95aee2bbb9677573bc4c23b78c4b5dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 05:17:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ffa48828b982e9731837e04e1156642130558ca7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 07:10:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1beac5aa2bbd9003099669f6b7763dd2b3d56e51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 07:10:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/830b23c713654bf7a49d2a7091910319 : SUCCESS in 1m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fa67d00be5346a9871913aa5d002a83 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/163299abdece4c67b5cc49b6cddd10a6 : SUCCESS in 3m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99977acf062d4fbcb8d5ca21893d4f9c : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/310226b66da04e7a99ac6459d72272cd : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/269b4943610649839c51c455ec93457b : SUCCESS in 8m 30s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/80f5baa6e56a48638d64ada6102b292e : SUCCESS in 1h 49m 19s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/cb0fe9d5c8674b7fad689bf3db270cea : SUCCESS in 1h 50m 54s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/74eed94f91bf4f43b03575cd85659b54 : SUCCESS in 1h 12m 26s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/d0d4c0145c48455990db836a46358a78 : SUCCESS in 1h 22m 53s","accounts_in_message":[],"_revision_number":1},{"id":"9ef945ef58b0b605b20a9613240ef9097a200fcf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 07:12:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e346e755f1234a0c8ef9996f15c404aa : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d61ca27745b09ad4a971c2354a9df386559d01a5","revisions":{"d61ca27745b09ad4a971c2354a9df386559d01a5":{"kind":"REWORK","_number":1,"created":"2020-08-26 14:34:24.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/46/748246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/46/748246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/46/748246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/46/748246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/46/748246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/46/748246/1"}}},"commit":{"parents":[{"commit":"156fed50fb7c9a930d7b7a8a27b5fb1030344c50","subject":"Merge \"Additional testing for Metal3 string cases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/156fed50fb7c9a930d7b7a8a27b5fb1030344c50"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-08-26 14:33:19.000000000","tz":480},"committer":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-08-26 14:34:19.000000000","tz":480},"subject":"Remove the unused coding style modules","message":"Remove the unused coding style modules\n\nPython modules related to coding style checks (listed in blacklist.txt in\nopenstack/requirements repo) are dropped from lower-constraints.txt\nthey are not needed during installation.\n\nChange-Id: I3a686b00cb2a6cb7cde4a26001d6d9016aab08eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/d61ca27745b09ad4a971c2354a9df386559d01a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/d61ca27745b09ad4a971c2354a9df386559d01a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
