)]}'
{"id":"openstack%2Fironic-lib~766815","triplet_id":"openstack%2Fironic-lib~stable%2Ftrain~If435ed0efe034d4bb99f0281ebc73840c0f38fb3","project":"openstack/ironic-lib","branch":"stable/train","hashtags":[],"change_id":"If435ed0efe034d4bb99f0281ebc73840c0f38fb3","subject":"Fix other out of memory error for qemu-img convert","status":"MERGED","created":"2020-12-14 18:51:23.000000000","updated":"2021-01-14 17:04:11.000000000","submitted":"2021-01-14 17:02:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"766815","cherry_pick_of_change":767007,"cherry_pick_of_patch_set":2,"meta_rev_id":"a4ac2606f9e1521efa1ca8b8c865042c34fcf95d","_number":766815,"virtual_id_number":766815,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-14 17:02:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2021-01-14 15:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-12-16 03:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2021-01-14 15:37:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-14 19:29:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-16 03:34:43.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"ee41c7a9d8f31e1707ac706357193f343c2a1241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-14 18:51:23.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"756cb88a9a0e8709a7d4165b80521664a162b9c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 19:29:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d66e2ed474349a089041758d68f0440 : SUCCESS in 6m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bf39028bde9430b84c824e16cae3048 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77ec5dbd23ee48faa4f18ae7e07c9ddb : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2dc5fe47e2c0410cb92f628d969d518a : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/938db2bdecd14dc291baee8e40b42766 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0df226a69f44b708a2d252f31eb0526 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8da3a911957d4c5da8f1edb1a7c9a6bc : SUCCESS in 5m 25s\n- ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/c23d5aad75dd4a45a55b2b5c2b32738d : FAILURE in 21m 12s\n- ironic-lib-partition-bios-ipmi-direct-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/ee0043393f31476b9e74a0cc00a86b97 : FAILURE in 22m 33s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/3d2f1288d0404c5e8a4afa56a04b17a3 : FAILURE in 26m 55s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/9b57c097ed0c433ab8ad11a366c272aa : FAILURE in 31m 53s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/50a4f2ebf1e24291932c0ab0cbfc6e83 : FAILURE in 29m 40s","accounts_in_message":[],"_revision_number":1},{"id":"0ee7f61743bf9a9cb785a832c4036a4f89e4e47e","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-12-16 03:34:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a58410c6aa8879bf305a7821f4d6863a47d65ef2","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-13 23:09:42.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ae355dedecd84387084a98a0c73e80457c63e4f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 00:40:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbd457ae34194f1592bb83502a299a96 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d95fcbc23f92472da92360a437e27856 : SUCCESS in 3m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50552afe21f24a6ea5b2f64e440f28c2 : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d7500e36d114e5cb46ef92db70197a9 : SUCCESS in 3m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f010f5a1a326411fad67704045a821ee : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01d34508f2a94133b2bd5a17679288ad : SUCCESS in 5m 13s\n- ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/e3136a38d3914e1eb99ae6a870fb742d : SUCCESS in 1h 14m 42s\n- ironic-lib-partition-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/7b4a7c8e7ed7454ca8466f8307acf497 : SUCCESS in 1h 12m 30s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/149bf73529da4ad5a2bb50c86dbabb09 : SUCCESS in 1h 14m 21s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/6de1f5159afe4d59993111ef93fc928d : SUCCESS in 1h 03m 14s","accounts_in_message":[],"_revision_number":1},{"id":"701a0cab72c3ba0ec14c555468b2a27adbc91f16","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-14 15:37:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8b7784d278ac7c73fa67c4d381ca35a0d986bd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 15:45:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f18180ed8f417843922f7f1673f8ff47c73012f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:02:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a07f02030352481092bc0f589a1a1b7e : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2aaf9e0b15c7498a97636946fd49219f : SUCCESS in 3m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3a25f0730e94ad8bfb7e62199696bfb : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c26743b8f8214e7ea44114aa0d0f3d1b : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/600ac4db23f549eb8e87ba7777c4bf26 : SUCCESS in 7m 01s\n- ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/7703e5a7ddfd427292b455a73b8bf4c9 : SUCCESS in 1h 13m 12s\n- ironic-lib-partition-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/fed237982c8c415db1420a6915bf270d : SUCCESS in 1h 13m 42s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/312309f94c15487395f30f57648b2a4b : SUCCESS in 1h 08m 14s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/ad685e6499f9452fa43ba5029ab642ea : SUCCESS in 1h 05m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a06d482dcc08d443565ac99d2e0d0a4368104f29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:02:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4ac2606f9e1521efa1ca8b8c865042c34fcf95d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:04:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a60399a4f3f421d87aea2cbdb3abc20 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a247e2b5bc3e87dbd0bf4b9409cf16321289741b","revisions":{"a247e2b5bc3e87dbd0bf4b9409cf16321289741b":{"kind":"REWORK","_number":1,"created":"2020-12-14 18:51:23.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/15/766815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/15/766815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/15/766815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/15/766815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/15/766815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/15/766815/1"}}},"commit":{"parents":[{"commit":"cc50dab51325bdf03fd840f7cb69bea1af2ea99f","subject":"Fix for latest zeroconf version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cc50dab51325bdf03fd840f7cb69bea1af2ea99f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-12-08 22:02:04.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-12-14 18:51:23.000000000","tz":0},"subject":"Fix other out of memory error for qemu-img convert","message":"Fix other out of memory error for qemu-img convert\n\nqemu-img convert makes a number of allocation requests, and less\ncommon than the \"Resource temporarily unavailable\" error is\n\"Cannot allocate memory\". The first seems to be largely raised\nupon a thread creation for the image conversion, where the first\nappears to be on buffer setup on the original process.\n\nRegardless, we should check for both and retry conversion\nautomatically.\n\nrhbz# 1892773\n\nNote: Because of a later patch which transformed the retry logic to\nleverage tenacity, this change has been modified accordingly.\n\nChange-Id: If435ed0efe034d4bb99f0281ebc73840c0f38fb3\n(cherry picked from commit 9be6afc751f0dbc251c4b5eeed71fd3b3659976c)\n(cherry picked from commit 9790280301ae2903906f2a583ccadec270029793)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/a247e2b5bc3e87dbd0bf4b9409cf16321289741b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/a247e2b5bc3e87dbd0bf4b9409cf16321289741b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
