)]}'
{"id":"openstack%2Fironic-lib~767203","triplet_id":"openstack%2Fironic-lib~stable%2Fvictoria~Iab279c57676dba8faa87ab7523e577d3b65eb428","project":"openstack/ironic-lib","branch":"stable/victoria","topic":"remove-l-c-job","hashtags":[],"change_id":"Iab279c57676dba8faa87ab7523e577d3b65eb428","subject":"Remove lower-constraints job","status":"MERGED","created":"2020-12-15 19:26:08.000000000","updated":"2020-12-18 03:23:02.000000000","submitted":"2020-12-18 03:16:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767203-remove-l-c-job","meta_rev_id":"3a2b462babd19237ccf7d465e9729c1ea5db88b7","_number":767203,"virtual_id_number":767203,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 03:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2020-12-17 01:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2020-12-17 16:19:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 19:26:08.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-12-15 20:43:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-17 01:29:16.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"}],"messages":[{"id":"07736b461284a6c9e2c1c1e376e5c428d5760b1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-15 19:26:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8ae732b899a264395784c39931a17c554519323","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 20:43:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f12d27683494d628efbc4da4a2754f6 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce7ef1a269b4211b98839774bbe17ee : SUCCESS in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f37e8da5ae3f4fbfb1710124cab61746 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54ed909952304510bff6bf2c979ff42d : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e92b465957344778febed6de1d05408 : SUCCESS in 5m 04s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/a4d17ace70864e579c1707fca67bf58f : FAILURE in 1h 03m 03s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/48da98b6526d4dbca89c102988799c4b : FAILURE in 1h 04m 32s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/c3ec3baecb424e138bcb2e4d9cbd900d : FAILURE in 55m 32s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/2d31ce26be534da2883396f6a633d26a : FAILURE in 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"3bac42a6ce89947a7548e248fffaaf38cace6250","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-16 15:49:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"14eb12b8bc43e79e311518312eeeb534b3f0c288","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 17:19:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/121208d380e94f19805adbad1b79c29a : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e151977cba4ab490f381f0812f4806 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c5c0d8198aa4a009f34b6f03d005fe8 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/252e644b219545e3af27d0b0a9f58328 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ced53e65bff14ed8a59df7541baa09ad : SUCCESS in 4m 38s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/4a2703ccb07644e193e905b66d2f6a69 : SUCCESS in 1h 20m 40s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/2e7dc39d1f77402b89ebc00caf2724cf : SUCCESS in 1h 21m 20s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/711f3c748d8440a696242ce11049627e : SUCCESS in 1h 16m 22s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/1937c72e23634a2693415370121212bd : SUCCESS in 1h 16m 41s","accounts_in_message":[],"_revision_number":2},{"id":"b1803d7792a86ee7ea37b4cc7b0984ac34601eb3","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2020-12-17 01:29:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d11713985f1f61f980071a03d00448c6332ac41c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-17 11:36:48.000000000","message":"Topic set to remove-l-c-job","accounts_in_message":[],"_revision_number":2},{"id":"fba91d35f08a3ac327be10415cca24354622bff1","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-17 16:19:46.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ee2c38ff2abca121117713005a5fe71f2e17f649","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 16:20:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2f84cb43273ddb88d8158f44495a380df5ffd566","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 18:28:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23be825584054ff79abf8c8e3f0a56f3 : SUCCESS in 2m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e2f13c06a1c46e78af464bdf5935754 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/679e34f69ac7425a84ffb8ab309299bb : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/282f0036eefc4488afa0c8bac41b62f1 : SUCCESS in 4m 31s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/4fc395b635fe47139c4b127961f01fe6 : TIMED_OUT in 2h 05m 11s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/cd17ea164ed74879a4a06e7a23cacad8 : SUCCESS in 1h 11m 29s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/132df42bd0094fd2bdfe0ece1adaf156 : SUCCESS in 1h 20m 04s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/c98b196a058a4d0a94bef9e045f1295f : SUCCESS in 1h 29m 22s","accounts_in_message":[],"_revision_number":2},{"id":"63ff8010e10f5fb6ec8811afa64320f6b3ccb948","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-17 18:53:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d50103229927c10cb563d27e5236be0787e7a937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 20:25:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a440c1519e3a46fe83ec0cfb4ffd6f2d : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c375f98c561648e2b7384bf9cd3f1284 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/449bca1598ab4a34b5a1a20231645a09 : SUCCESS in 2m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73ccd5139b234fa888f9a73e6303275d : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff53112f724a432c9c2e6d4958b41418 : SUCCESS in 5m 08s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/9a065f75b3c441f7ac86f939271f276f : SUCCESS in 1h 22m 02s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/72573f4645fc43bb9f7d96c22b02aaa0 : SUCCESS in 1h 14m 16s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/f9c1ec2b4d134091ad356aa93f9f1dc6 : SUCCESS in 1h 27m 07s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/4577c9580a434d6d811309502bec7c97 : SUCCESS in 1h 26m 33s","accounts_in_message":[],"_revision_number":2},{"id":"1db1d54de5e7236a8d30aaff809dbe4cca29049c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 20:26:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"191d12250835db31e163b41b8dc4729018fb2e93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 03:16:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f92639366164d3c95d0ee131f249280 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f21e08e5e97485eb0eced8195967560 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b701fee9e214724af43120ea8ba01b6 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7003daa705c9442ebbe538d223c3c78e : SUCCESS in 5m 47s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/baf8dfb87ab74d2eb473d6d55029272a : SUCCESS in 1h 30m 11s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/20c0dd578fed40299e43651c3dd00ce3 : SUCCESS in 1h 27m 26s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/4aa3a18854e647cf890592781d624da9 : SUCCESS in 1h 27m 45s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/8e74487b38374c21a646da47688d462f : SUCCESS in 1h 26m 47s","accounts_in_message":[],"_revision_number":2},{"id":"1f8c78a6930e99f6af349383a8d4f684d48f9e20","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 03:16:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3a2b462babd19237ccf7d465e9729c1ea5db88b7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 03:23:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19112da06ff94e54961abdf64d5f79d6 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"aa7cfec18c616c0ed1d1a70ae8aee8696a22c7c9","revisions":{"d5ae74e9b8a6b2aa602095b60286d5ab2f1da63a":{"kind":"REWORK","_number":1,"created":"2020-12-15 19:26:08.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/03/767203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/03/767203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/03/767203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/03/767203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/03/767203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/03/767203/1"}}},"commit":{"parents":[{"commit":"e1b5e12cb1636178835d0d9137045e9afcf5ff1a","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/e1b5e12cb1636178835d0d9137045e9afcf5ff1a"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-14 16:24:18.000000000","tz":60},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-12-15 19:25:41.000000000","tz":-480},"subject":"Remove lower-constraints job","message":"Remove lower-constraints job\n\nAs discussed during the upstream ironic community meeting on\nMonday Dec 14 2020, the lower-constraints job is being removed.\n\nChange-Id: Iab279c57676dba8faa87ab7523e577d3b65eb428\n(cherry picked from commit 0bbba8e62f0b9f68df5098c31a0eec6b428c2797)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/d5ae74e9b8a6b2aa602095b60286d5ab2f1da63a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/d5ae74e9b8a6b2aa602095b60286d5ab2f1da63a"}]},"branch":"refs/heads/stable/victoria"},"aa7cfec18c616c0ed1d1a70ae8aee8696a22c7c9":{"kind":"REWORK","_number":2,"created":"2020-12-16 15:49:18.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/03/767203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/03/767203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/03/767203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/03/767203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/03/767203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/03/767203/2"}}},"commit":{"parents":[{"commit":"e1b5e12cb1636178835d0d9137045e9afcf5ff1a","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/e1b5e12cb1636178835d0d9137045e9afcf5ff1a"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-14 16:24:18.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-16 15:48:52.000000000","tz":60},"subject":"Remove lower-constraints job","message":"Remove lower-constraints job\n\nAs discussed during the upstream ironic community meeting on\nMonday Dec 14 2020, the lower-constraints job is being removed.\n\nAlso disabling dstat in all focal based jobs because of\nhttps://bugs.launchpad.net/ubuntu/+source/dstat/+bug/1866619\n\nChange-Id: Iab279c57676dba8faa87ab7523e577d3b65eb428\n(cherry picked from commit 0bbba8e62f0b9f68df5098c31a0eec6b428c2797)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/aa7cfec18c616c0ed1d1a70ae8aee8696a22c7c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/aa7cfec18c616c0ed1d1a70ae8aee8696a22c7c9"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
