)]}'
{"id":"openstack%2Fironic-lib~767207","triplet_id":"openstack%2Fironic-lib~stable%2Ftrain~Iab279c57676dba8faa87ab7523e577d3b65eb428","project":"openstack/ironic-lib","branch":"stable/train","topic":"remove-l-c-job","hashtags":[],"change_id":"Iab279c57676dba8faa87ab7523e577d3b65eb428","subject":"CI: Use only py3, remove lower-constraints check, cap ipa-b version","status":"MERGED","created":"2020-12-15 19:30:30.000000000","updated":"2020-12-18 03:23:10.000000000","submitted":"2020-12-18 03:16:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767207-remove-l-c-job","meta_rev_id":"9bfec658d81725c291c9da496013c2cd48d8e89b","_number":767207,"virtual_id_number":767207,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-12-18 03:16:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2020-12-17 22:11:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2020-12-17 22:11:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 19:30:30.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2020-12-15 20:26:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-17 22:11:07.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"9e64f4c59aec31034e50b755191c4977fe8048fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-15 19:30:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08e094d3855c4868ddd46bbdbafea16f3f8b5350","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 20:26:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cd6583343d2417cb3b9b4dd07106446 : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/280aea07e28c4a09972c213e9a44f443 : SUCCESS in 4m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32c9a23662c34e959526d34847bfd2df : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c857251ef064da1b0bf17eadf499b7c : SUCCESS in 3m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ccb79da36333450cb764b58b7971f9cf : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d068ae04818f4c47b7d27e257d650558 : SUCCESS in 4m 34s\n- ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/d425d3aa00a543f78ccab659dad75bd2 : FAILURE in 18m 40s\n- ironic-lib-partition-bios-ipmi-direct-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/63dc79711f944aa8914282c1c28693e6 : FAILURE in 17m 16s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/7852c754da474c8a8ab2179fbe092ff7 : FAILURE in 34m 59s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/4bb5397dcfcc495780e76491802e28d1 : FAILURE in 30m 16s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/f07f444fa12f49c4835707b6e1a3417e : FAILURE in 32m 13s","accounts_in_message":[],"_revision_number":1},{"id":"feecb40e00f9a5dfeb3c182431fd8ed3930ab751","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2020-12-17 11:36:55.000000000","message":"Topic set to remove-l-c-job","accounts_in_message":[],"_revision_number":1},{"id":"873f57b1c92c8561779ef9e104269d85165a1b35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-17 19:03:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7c4703cf1da30042efcb6f6d35cac8757e57db4f","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-17 19:03:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b9f354664b708da3845c6ae698605003fb984686","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-17 19:14:52.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"f3a035f6129eb8f18de4939466a24accdbe74e8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 20:47:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8313a3c888e47e9a21f5c844a5c2b17 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62e73df2f8c84003b0fc8b8459df8b08 : SUCCESS in 2m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/49f729d33f2d4047af4eeb7323d003c1 : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3677220e0bd54452bc330223d3fc4f78 : SUCCESS in 3m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1c105727ebbb440aaf3781aa1f7f61d0 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e5af7e1f184294b53f8eb1c52be853 : SUCCESS in 16m 12s\n- ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/d6ed7e4dfdcc42e982dd7df6ce68dec0 : SUCCESS in 1h 08m 12s\n- ironic-lib-partition-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/dec28ee6a9c54f0dbfb5c5cf3fce463e : SUCCESS in 1h 25m 38s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/4cf1173d2bb647a8b7de8b34d336c5b6 : SUCCESS in 1h 10m 03s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/555ff90b2b27431baa3cadbf52046f7a : SUCCESS in 1h 21m 33s","accounts_in_message":[],"_revision_number":3},{"id":"ea4fd0b1e06cd4490db4563688190b54470adde4","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-12-17 22:11:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"137b24a2751099acb9e3dec26d72267e2479636e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 22:11:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2e6f4167e90ddaabf3b782084766d180d4dac982","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 03:16:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/908a561591b84984b8bbd36a3ee3ba66 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/609e09e3ebc4452dabfb4630d4cf9c3e : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1d53ba17e824c16837c21b003198c8d : SUCCESS in 3m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e90576d8873444a6abc0ffbaf28c39f4 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f74cb70c26964539977f01014406f22a : SUCCESS in 4m 50s\n- ironic-lib-wholedisk-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/0426c2a79de0496ebe829b7988d36b1e : SUCCESS in 1h 07m 54s\n- ironic-lib-partition-bios-ipmi-direct-tinyipa-src https://zuul.opendev.org/t/openstack/build/d2f7fb3eaeb04202bb3763b8d6f324ec : SUCCESS in 1h 06m 54s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/3d90a4b6963946f4b9975cb758da4bde : SUCCESS in 1h 03m 10s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/eec4f680373a431c8a585892d891a09b : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":3},{"id":"0654ec42f989e00ef2e10d02c3e3200cc162838b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 03:16:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9bfec658d81725c291c9da496013c2cd48d8e89b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 03:23:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff2dc38729b74847ab4e0da2ddaf9405 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"00d8ad3217e4bce5306b4639db034d1ea003cd7b","revisions":{"e0bf1351f7185e87f423fe3cf323b7945818e7eb":{"kind":"REWORK","_number":1,"created":"2020-12-15 19:30:30.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/07/767207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/07/767207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/07/767207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/07/767207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/07/767207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/07/767207/1"}}},"commit":{"parents":[{"commit":"cc50dab51325bdf03fd840f7cb69bea1af2ea99f","subject":"Fix for latest zeroconf version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cc50dab51325bdf03fd840f7cb69bea1af2ea99f"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-14 16:24:18.000000000","tz":60},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-12-15 19:30:20.000000000","tz":-480},"subject":"Remove lower-constraints job","message":"Remove lower-constraints job\n\nAs discussed during the upstream ironic community meeting on\nMonday Dec 14 2020, the lower-constraints job is being removed.\n\nChange-Id: Iab279c57676dba8faa87ab7523e577d3b65eb428\n(cherry picked from commit 0bbba8e62f0b9f68df5098c31a0eec6b428c2797)\n(cherry picked from commit d5ae74e9b8a6b2aa602095b60286d5ab2f1da63a)\n(cherry picked from commit 16ac91e9554ecdb0d8f55a42ab2bd8756a78671a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/e0bf1351f7185e87f423fe3cf323b7945818e7eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/e0bf1351f7185e87f423fe3cf323b7945818e7eb"}]},"branch":"refs/heads/stable/train"},"0baa7afb57a53d65eb923bdc9838f1e815e6a03e":{"kind":"REWORK","_number":2,"created":"2020-12-17 19:03:20.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/07/767207/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/07/767207/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/07/767207/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/07/767207/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/07/767207/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/07/767207/2"}}},"commit":{"parents":[{"commit":"cc50dab51325bdf03fd840f7cb69bea1af2ea99f","subject":"Fix for latest zeroconf version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cc50dab51325bdf03fd840f7cb69bea1af2ea99f"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-14 16:24:18.000000000","tz":60},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-12-17 19:01:27.000000000","tz":-480},"subject":"CI: Use only py3, remove lower-constraints check, cap ipa-b version","message":"CI: Use only py3, remove lower-constraints check, cap ipa-b version\n\nAs discussed during the upstream ironic community meeting on\nMonday Dec 14 2020, the lower-constraints job is being removed.\n\nAlong with this, in order to unblock CI, we need to also cap the\nversion of ipa-b because the tinycore 11 image does not build on\nolder versions of linux, and along with this the testing\nlibrary constraints that are resolved now now error on python2 being\nused, so changing jobs to use only python3.\n\nChange-Id: Iab279c57676dba8faa87ab7523e577d3b65eb428\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/0baa7afb57a53d65eb923bdc9838f1e815e6a03e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/0baa7afb57a53d65eb923bdc9838f1e815e6a03e"}]},"branch":"refs/heads/stable/train"},"00d8ad3217e4bce5306b4639db034d1ea003cd7b":{"kind":"REWORK","_number":3,"created":"2020-12-17 19:14:52.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/07/767207/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/07/767207/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/07/767207/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/07/767207/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/07/767207/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/07/767207/3"}}},"commit":{"parents":[{"commit":"cc50dab51325bdf03fd840f7cb69bea1af2ea99f","subject":"Fix for latest zeroconf version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cc50dab51325bdf03fd840f7cb69bea1af2ea99f"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2020-12-14 16:24:18.000000000","tz":60},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-12-17 19:14:48.000000000","tz":0},"subject":"CI: Use only py3, remove lower-constraints check, cap ipa-b version","message":"CI: Use only py3, remove lower-constraints check, cap ipa-b version\n\nAs discussed during the upstream ironic community meeting on\nMonday Dec 14 2020, the lower-constraints job is being removed.\n\nAlong with this, in order to unblock CI, we need to also cap the\nversion of ipa-b because the tinycore 11 image does not build on\nolder versions of linux, and along with this the testing\nlibrary constraints that are resolved now now error on python2 being\nused, so changing jobs to use only python3.\n\nChange-Id: Iab279c57676dba8faa87ab7523e577d3b65eb428\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/00d8ad3217e4bce5306b4639db034d1ea003cd7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/00d8ad3217e4bce5306b4639db034d1ea003cd7b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
