)]}'
{"id":"openstack%2Fironic-lib~767221","triplet_id":"openstack%2Fironic-lib~stable%2Ftrain~I58fcaf690d62f5c00a7f294c26a0612633ccc2ce","project":"openstack/ironic-lib","branch":"stable/train","hashtags":[],"change_id":"I58fcaf690d62f5c00a7f294c26a0612633ccc2ce","subject":"Use python3 for stable/train testing","status":"ABANDONED","created":"2020-12-15 21:12:02.000000000","updated":"2020-12-17 22:27:36.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"53bfe373de80c32e966ee68052e6d8813291b09d","_number":767221,"virtual_id_number":767221,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-12-16 22:06:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-15 21:49:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-16 03:18:59.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"CC"},{"updated":"2020-12-16 03:19:39.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"}],"messages":[{"id":"ba31a0ae6c43aca256435f45ff98da4f7b295fbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-15 21:12:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"82fb059b6b187a91a64a8695c37579d2b9bf7bbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 21:49:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1e75431ff4f46eda77f4529d3bbc55d : SUCCESS in 5m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aefee5adf1d246e1ac762e5ace1ac75d : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7dc09aa4ca2429db4280dcbd8d58be5 : SUCCESS in 3m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6531da93055549e7867a736788bc0138 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ee015c823d24a65b0b5665baf40628c : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/523ba11e7f2249699c36745dfb6563bf : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa9d1513e65446fafc503752177fc4a : SUCCESS in 4m 51s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src-python2 https://zuul.opendev.org/t/openstack/build/d7523fa5455b48508964f0b187a8b415 : FAILURE in 29m 23s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/bb6b1ccbde0c481588c5a81bff5313b1 : FAILURE in 24m 43s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/615c6bf50b6a4679884b94a753a08e19 : FAILURE in 28m 01s","accounts_in_message":[],"_revision_number":1},{"id":"2941d7f76e07effba77e7fc6236f9301252eb1f9","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-12-16 03:18:59.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nI\u0027m going to try aligning this with what stable/ussuri does","accounts_in_message":[],"_revision_number":1},{"id":"60c74b47cbadf7cb18286cbc00353771f4d8736f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-12-16 03:19:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"77a04bba69de858947a29d05e995010ce79492ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 04:07:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e56dd6130eb4f8a85901cec2abe8304 : SUCCESS in 3m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12519af018ce451b80257f3d74668325 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d485907fd70746dd9edc1d08a92baabf : SUCCESS in 3m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0583433a65a749cb8dd80268c78e59aa : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd12909ee1f44eeebd611add645e5117 : SUCCESS in 4m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c20e03d660874fe19241793120b0ff69 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/037f2448e4ef4f38b5cb213fadbb93c8 : SUCCESS in 5m 04s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/b821af711976469f973e7dac205dbe8d : FAILURE in 38m 25s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/629647bb13614d8e83c3996ca53706e8 : FAILURE in 29m 18s","accounts_in_message":[],"_revision_number":2},{"id":"730a48b1a3839d992df8e390a6feb24b442c0800","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-12-16 19:47:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1a530f93ea075d424d745a4041efd4f804bdbb42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-12-16 20:14:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"adc3477e69404a1ffc62bfa172c9c9333a2e714f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 20:32:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e7eaecb19c34b268fe0e52ebaa6b92e : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4ee1c78528ec4645b48dbe4585545e08 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1eba741a3c1141a69cc5457044246bf4 : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f742829bf97b4d758ac2dc79b1e4d022 : SUCCESS in 3m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/534bf7b465ea42ba9f32d2221397a2ab : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7fd539eb198e4c289aaf785445c6e930 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ab7bc0afe38429a93785e5a3d956ae6 : SUCCESS in 4m 34s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/3f8858185fc348d8a3ad17e2da084225 : FAILURE in 10m 47s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/d148e3889f2b4cb69fb1fd645cbdf512 : FAILURE in 11m 00s","accounts_in_message":[],"_revision_number":3},{"id":"d3e6b8df8e24e09af4c7c4ac99ffff4d007520f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-12-16 21:33:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5413f2ba62011a50cac8ac4ff9ab51fef5a641f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 22:06:54.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30ce00e21fe24333a6a2b3c6379bc521 : SUCCESS in 5m 26s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4aa87ae3367a4647afca539178f11ab3 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a312052be765421b86974454c7b49c71 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b314f7225a284732a1911a23ce49c6ee : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b586f0853ea4e46908cc19e3c91a0ad : SUCCESS in 3m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6962fba1c3e34adc9e9c3844e2dc1b85 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3aaea5ad97242d788a2801415666c4f : SUCCESS in 5m 23s\n- ironic-lib-partition-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/5c7b0ce8b40d4d4b8f752195884497e7 : FAILURE in 29m 32s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-tinyipa-src https://zuul.opendev.org/t/openstack/build/3f63bef45f9640e3a0362038cab8154d : FAILURE in 28m 25s","accounts_in_message":[],"_revision_number":4},{"id":"53bfe373de80c32e966ee68052e6d8813291b09d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2020-12-17 22:27:36.000000000","message":"Abandoned\n\nSuperseded by https://review.opendev.org/c/openstack/ironic-lib/+/767207","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"f483f2f75f94ccb27ff1920bd99db95d3dddef5d","revisions":{"ff1e2507332b8ff8c6c074fedaa788ffdb3bea42":{"kind":"REWORK","_number":1,"created":"2020-12-15 21:12:02.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/21/767221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/21/767221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/1"}}},"commit":{"parents":[{"commit":"cc50dab51325bdf03fd840f7cb69bea1af2ea99f","subject":"Fix for latest zeroconf version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cc50dab51325bdf03fd840f7cb69bea1af2ea99f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-12-15 21:06:38.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-12-15 21:11:38.000000000","tz":-480},"subject":"Use python3 for stable/train testing","message":"Use python3 for stable/train testing\n\nWe have a dependency issue:\n\nbandit requires Python \u0027\u003e\u003d3.5\u0027 but the running Python is 2.7.17\n\nHowever, python2.7 support is no longer required and we need\ntesting generally working to validate fixes are in a decent state.\n\nChange-Id: I58fcaf690d62f5c00a7f294c26a0612633ccc2ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/ff1e2507332b8ff8c6c074fedaa788ffdb3bea42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/ff1e2507332b8ff8c6c074fedaa788ffdb3bea42"}]},"branch":"refs/heads/stable/train"},"553def5d7b576fd1d260d8a638fa68c7fa9b318f":{"kind":"REWORK","_number":2,"created":"2020-12-16 03:19:39.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/21/767221/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/21/767221/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/2"}}},"commit":{"parents":[{"commit":"cc50dab51325bdf03fd840f7cb69bea1af2ea99f","subject":"Fix for latest zeroconf version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cc50dab51325bdf03fd840f7cb69bea1af2ea99f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-12-15 21:06:38.000000000","tz":-480},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-12-16 03:19:27.000000000","tz":780},"subject":"Use python3 for stable/train testing","message":"Use python3 for stable/train testing\n\nWe have a dependency issue:\n\nbandit requires Python \u0027\u003e\u003d3.5\u0027 but the running Python is 2.7.17\n\nHowever, python2.7 support is no longer required and we need\ntesting generally working to validate fixes are in a decent state.\n\nChange-Id: I58fcaf690d62f5c00a7f294c26a0612633ccc2ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/553def5d7b576fd1d260d8a638fa68c7fa9b318f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/553def5d7b576fd1d260d8a638fa68c7fa9b318f"}]},"branch":"refs/heads/stable/train"},"a4b4b02e3e031d1fca4a14c614aa29c704c4cddc":{"kind":"REWORK","_number":3,"created":"2020-12-16 20:14:26.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/21/767221/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/21/767221/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/3"}}},"commit":{"parents":[{"commit":"cc50dab51325bdf03fd840f7cb69bea1af2ea99f","subject":"Fix for latest zeroconf version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cc50dab51325bdf03fd840f7cb69bea1af2ea99f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-12-15 21:06:38.000000000","tz":-480},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-12-16 20:13:33.000000000","tz":780},"subject":"Use python3 for stable/train testing","message":"Use python3 for stable/train testing\n\nWe have a dependency issue:\n\nbandit requires Python \u0027\u003e\u003d3.5\u0027 but the running Python is 2.7.17\n\nHowever, python2.7 support is no longer required and we need\ntesting generally working to validate fixes are in a decent state.\n\nThis change also switches to focal due to \"FATAL: kernel too old\"\nerror.\n\nChange-Id: I58fcaf690d62f5c00a7f294c26a0612633ccc2ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/a4b4b02e3e031d1fca4a14c614aa29c704c4cddc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/a4b4b02e3e031d1fca4a14c614aa29c704c4cddc"}]},"branch":"refs/heads/stable/train"},"f483f2f75f94ccb27ff1920bd99db95d3dddef5d":{"kind":"REWORK","_number":4,"created":"2020-12-16 21:33:52.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/21/767221/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/21/767221/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/21/767221/4"}}},"commit":{"parents":[{"commit":"cc50dab51325bdf03fd840f7cb69bea1af2ea99f","subject":"Fix for latest zeroconf version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cc50dab51325bdf03fd840f7cb69bea1af2ea99f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2020-12-15 21:06:38.000000000","tz":-480},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2020-12-16 21:33:33.000000000","tz":780},"subject":"Use python3 for stable/train testing","message":"Use python3 for stable/train testing\n\nWe have a dependency issue:\n\nbandit requires Python \u0027\u003e\u003d3.5\u0027 but the running Python is 2.7.17\n\nHowever, python2.7 support is no longer required and we need\ntesting generally working to validate fixes are in a decent state.\n\nChange-Id: I58fcaf690d62f5c00a7f294c26a0612633ccc2ce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/f483f2f75f94ccb27ff1920bd99db95d3dddef5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/f483f2f75f94ccb27ff1920bd99db95d3dddef5d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
