)]}'
{"id":"openstack%2Fironic-lib~771788","triplet_id":"openstack%2Fironic-lib~master~Ie9809b3e4fe2bea6b66ee95c442111942e85d618","project":"openstack/ironic-lib","branch":"master","topic":"story_2008539","hashtags":[],"change_id":"Ie9809b3e4fe2bea6b66ee95c442111942e85d618","subject":"Handle EBRs and tiny partitions when removing metadata","status":"MERGED","created":"2021-01-21 10:41:34.000000000","updated":"2021-01-27 13:46:33.000000000","submitted":"2021-01-27 13:35:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"771788-story_2008539","meta_rev_id":"b76698d4cd51ad53357161f148151fb3d696abeb","_number":771788,"virtual_id_number":771788,"owner":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-27 13:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-27 12:01:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":2,"date":"2021-01-25 18:51:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-27 12:01:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-21 12:30:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-24 09:42:13.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-01-27 12:01:17.000000000","updated_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"reviewer":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"state":"REVIEWER"}],"messages":[{"id":"55d1ce12a72379dd053a3e481520833d65234448","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-01-21 10:41:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3504bd33b5c4c880e38b89136469ea2fa7ca6892","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-21 12:30:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2519b92b2f284dc289246b759c85de16 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/894820da900941d2b7b8068eee6e0672 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e2b950e0b0c475e9ad1a89c91b1d6b2 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b5e7fef5f1449e29a81c2f5206868d6 : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0105b59104db47fca659e1767eeaa328 : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3e034d06e4541e99bf324596559a8c3 : SUCCESS in 5m 50s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/f3dd6869052045dda856786073c8068b : SUCCESS in 1h 23m 08s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/1c86e42d0bef4fd7989bf884ac189783 : SUCCESS in 1h 15m 21s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/2c1addb8b53146899faf6f5985616d23 : SUCCESS in 1h 13m 21s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/bd8547060c3d49e4a8ed88fc4dd631b0 : SUCCESS in 1h 45m 53s","accounts_in_message":[],"_revision_number":1},{"id":"ac8bd98865842eabdfed6fd82092a84cacea0a1a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-01-24 09:42:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nCould you add/update unit tests?","accounts_in_message":[],"_revision_number":1},{"id":"9b38504a8e2a146f249df920e8812400509bf87c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-01-25 17:59:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e99e63cfe84570fb5a7c6c860cc63a9954437723","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-01-25 18:01:37.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Could you add/update unit tests?\n\nThanks for the review, Dmitry. Unit tests added.","accounts_in_message":[],"_revision_number":2},{"id":"562cb917b7fe6d26717239d3a6fa7f24fc7f8937","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-01-25 18:51:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8d3ed75d51083d243baa1b1681123a776dfd925d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 20:20:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f40d6b23fe343b894c084c825cfae73 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b0c4f4713f647ffac0201d2645eb7b1 : SUCCESS in 2m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aef69ce448274bba8e85ddf41c0d2226 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50c99ee6e71a466db064e2ee5c0afae6 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c63500f63738444ea734e043493220a7 : SUCCESS in 3m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fbcca351b55480cb91b5c1b37949e3f : SUCCESS in 6m 03s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/6618067d66cf45d0b82506878b22008e : SUCCESS in 1h 55m 25s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/779a737713ec40e8819256ccbb09410f : SUCCESS in 1h 23m 01s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/d12b7b7bfbcc4cffbbb76104aff638ca : SUCCESS in 1h 26m 45s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/2bcdcfe60afd473687e52884aef2a920 : TIMED_OUT in 2h 03m 58s","accounts_in_message":[],"_revision_number":2},{"id":"473690aa707210338acddee80b259ca3b887d869","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-01-26 09:00:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a7271d13f3ec510f9ced5551fbc9daf41298864d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:36:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e729a13b884455fb7eb1f7f68b10e35 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e5d2e28522d4990b2a232a9c05db813 : SUCCESS in 3m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ca28f86250541a88e1429d2a75ba710 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3159bb4245234b6baef84764fd05911c : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af880798a4b842f0986742d80246e64b : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/787771b96d4045ef97fe4b74af12dc33 : SUCCESS in 5m 27s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/ea65c0ed2b534f8796c42ddfeed1fe22 : SUCCESS in 1h 24m 40s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/1e91c1ef323d4a2c953ec62de70bc7c3 : SUCCESS in 1h 32m 58s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/557c43c7eac34c57a266675592a8dde6 : SUCCESS in 1h 15m 21s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/cfbb9d9ecdd64bb589c8e01bfdfe6ca7 : SUCCESS in 1h 29m 01s","accounts_in_message":[],"_revision_number":2},{"id":"95107df571a13163bbf29f2fa4fce4dfed5e9be9","author":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"},"date":"2021-01-27 12:01:17.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThank you Arne, hit a same issue here.","accounts_in_message":[],"_revision_number":2},{"id":"ebe99ce0b5d438986cd014c53a0327b523081ca2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 12:01:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3cc10a75ca4f95b50f799e9839e540c0eb982a27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 13:35:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d48c3c8aff9049baacdf37bd33643a87 : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38d9c77b289c46c68a24489a99f64ad0 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/435f000106a84bea9ee117c10a21772c : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fab0eccdc0be453581d7dda47aa70877 : SUCCESS in 6m 58s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/e2fb1e34340e4957a6dc839a53f7a57d : SUCCESS in 1h 14m 31s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/95502e85daa8406cb205ff5d5f541b74 : SUCCESS in 1h 27m 34s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/21086bde79174811b3d091b6c192abde : SUCCESS in 1h 13m 04s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/36154eb9ccfc471d825e490f0afc3583 : SUCCESS in 1h 26m 31s","accounts_in_message":[],"_revision_number":2},{"id":"11f13c4cd9f81e129070cdc6a92843eff028e06e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 13:35:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b76698d4cd51ad53357161f148151fb3d696abeb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 13:46:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e56dd154d7994c6a864c3a6f47aacc1b : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9a60d3853fc5b36cd3715e9419568cc9c51f5a8e","revisions":{"a48802aa64424dc620582dd61cac6cb7b9160fcd":{"kind":"REWORK","_number":1,"created":"2021-01-21 10:41:34.000000000","uploader":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"ref":"refs/changes/88/771788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/88/771788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/88/771788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/88/771788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/88/771788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/88/771788/1"}}},"commit":{"parents":[{"commit":"ebdf6784cbe22f4cbe76229a8fd36a92acf35835","subject":"Import mounted context manager from IPA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/ebdf6784cbe22f4cbe76229a8fd36a92acf35835"}]}],"author":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2021-01-20 18:03:07.000000000","tz":60},"committer":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2021-01-21 10:09:20.000000000","tz":60},"subject":"Handle EBRs and tiny partitions when removing metadata","message":"Handle EBRs and tiny partitions when removing metadata\n\nExtended boot records (EBRs) are used in the MSDOS partitioning\nscheme to manage extended partitions. EBRs are small partitions\nthemselves and Ironic will therefore treat them like any other\npartition during cleaning. Since EBRs are smaller than a GPT\n(33 sectors), dd\u0027ing 33 sectors will fail. This patch proposes\na fix to handle EBRs (and other tiny paritions) by adjusting\nthe number of blocks writtent to be the partition size at max.\n\nStory: #2008539\nTask: #41628\nChange-Id: Ie9809b3e4fe2bea6b66ee95c442111942e85d618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/a48802aa64424dc620582dd61cac6cb7b9160fcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/a48802aa64424dc620582dd61cac6cb7b9160fcd"}]},"branch":"refs/heads/master"},"9a60d3853fc5b36cd3715e9419568cc9c51f5a8e":{"kind":"REWORK","_number":2,"created":"2021-01-25 17:59:57.000000000","uploader":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"ref":"refs/changes/88/771788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/88/771788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/88/771788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/88/771788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/88/771788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/88/771788/2"}}},"commit":{"parents":[{"commit":"ebdf6784cbe22f4cbe76229a8fd36a92acf35835","subject":"Import mounted context manager from IPA","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/ebdf6784cbe22f4cbe76229a8fd36a92acf35835"}]}],"author":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2021-01-20 18:03:07.000000000","tz":60},"committer":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2021-01-25 17:58:56.000000000","tz":60},"subject":"Handle EBRs and tiny partitions when removing metadata","message":"Handle EBRs and tiny partitions when removing metadata\n\nExtended boot records (EBRs) are used in the MSDOS partitioning\nscheme to manage extended partitions. EBRs are small partitions\nthemselves and Ironic will therefore treat them like any other\npartition during cleaning. Since EBRs are smaller than a GPT\n(33 sectors), dd\u0027ing 33 sectors will fail. This patch proposes\na fix to handle EBRs (and other tiny paritions) by adjusting\nthe number of blocks writtent to be the partition size at max.\n\nStory: #2008539\nTask: #41628\nChange-Id: Ie9809b3e4fe2bea6b66ee95c442111942e85d618\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/9a60d3853fc5b36cd3715e9419568cc9c51f5a8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/9a60d3853fc5b36cd3715e9419568cc9c51f5a8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24828,"name":"Kaifeng Wang","email":"kaifeng.w@gmail.com","username":"wangkf"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
