)]}'
{"id":"openstack%2Fironic-lib~774354","triplet_id":"openstack%2Fironic-lib~master~I0e790cebcd086d26bf27b909f73842c3f46bbd77","project":"openstack/ironic-lib","branch":"master","topic":"json-rpc","hashtags":[],"change_id":"I0e790cebcd086d26bf27b909f73842c3f46bbd77","subject":"Import json_rpc from ironic","status":"MERGED","created":"2021-02-06 12:36:15.000000000","updated":"2021-02-28 14:43:07.000000000","submitted":"2021-02-28 14:41:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"774354-json-rpc","meta_rev_id":"c8dea909f999dbc8ab079abc6c1510daf23c72de","_number":774354,"virtual_id_number":774354,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-28 14:41:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-02-24 10:22:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-02-28 13:13:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-28 13:13:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-06 14:23:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-24 10:22:07.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-02-28 13:13:42.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"35d616c078d19376426fbbdb526b90d789b4b7a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-02-06 12:36:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5be7c2f9178998f133f423eb987c0b48ba3891c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-02-06 12:51:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"997f1699d1d8a75645ee126abdd69cff02c54597","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 14:23:16.000000000","message":"Patch Set 2: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/de1722b0e53f48028d88f86439ab2134 : SUCCESS in 1m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5ca1dbcc49564b6ab0fbf25403e64724 : SUCCESS in 1m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/209b9e14a36b4f0ca154ac72e9ada014 : FAILURE in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/305553aebc5d4660933d4703da04a207 : FAILURE in 2m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2239722056ae404ea979eb8a9d2da3ee : FAILURE in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/853673e1b7ff430fb179288bd810d9fb : FAILURE in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7c544d8a9c84dbcbe6514b9216b7cb2 : FAILURE in 3m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14aa5774763f4978962acb658e12c6ed : FAILURE in 3m 58s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/c38473e690734ab58d777ab710b6e805 : SUCCESS in 1h 27m 35s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/73ab3b544d5b472f9bca414932d1b3d5 : SUCCESS in 1h 30m 12s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/e1f7011895044a47b142fd5c00782984 : SUCCESS in 1h 28m 31s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/2c0c281ab1f14d748fb01d277583225a : SUCCESS in 1h 26m 01s","accounts_in_message":[],"_revision_number":2},{"id":"f63c642d2095f021d5c3f408752461d8aaddcf8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-02-08 12:43:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3e9029d380883eb769a6830971ebc785ff35f553","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 14:35:54.000000000","message":"Patch Set 3: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/129f5aa147f3487d96a5fabaf0644af7 : FAILURE in 2m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bb604911a4a74340aa77e1d4c99e7073 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/858db8587f6c445396fad7d3634d366d : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abecd5d8929b46749edbe21152f854c2 : FAILURE in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1b100789ce1438e8c57a803674c9212 : FAILURE in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a71ecfcfb534190a4d29360e68000c5 : FAILURE in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/621406f0ea9948f4a3dc4f0cd9731b0e : FAILURE in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b95c38fcf2b459ab11eca8185f8318c : SUCCESS in 5m 22s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/91c062f39afe4782af2b31dbf68db3d9 : SUCCESS in 1h 18m 18s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/d42d1e0bf9804d1199d5aad1799dd390 : SUCCESS in 1h 17m 19s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/fc17b5b6819441c1b775c3167a261c0d : FAILURE in 31m 11s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/5784c7e208f34337b7a583e1a1402d79 : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":3},{"id":"ea3cbf476a4ec6ec3764cc58f7c51a8ec92df2ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 14:36:34.000000000","message":"Patch Set 3:\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/129f5aa147f3487d96a5fabaf0644af7 : FAILURE in 2m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bb604911a4a74340aa77e1d4c99e7073 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/858db8587f6c445396fad7d3634d366d : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abecd5d8929b46749edbe21152f854c2 : FAILURE in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1b100789ce1438e8c57a803674c9212 : FAILURE in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a71ecfcfb534190a4d29360e68000c5 : FAILURE in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/621406f0ea9948f4a3dc4f0cd9731b0e : FAILURE in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b95c38fcf2b459ab11eca8185f8318c : SUCCESS in 5m 22s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/91c062f39afe4782af2b31dbf68db3d9 : SUCCESS in 1h 18m 18s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/d42d1e0bf9804d1199d5aad1799dd390 : SUCCESS in 1h 17m 19s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/fc17b5b6819441c1b775c3167a261c0d : FAILURE in 31m 11s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/5784c7e208f34337b7a583e1a1402d79 : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":3},{"id":"6eaf36fcb8970f7cf6c8e234b4bd16800abc87e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 14:37:24.000000000","message":"Patch Set 3:\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/129f5aa147f3487d96a5fabaf0644af7 : FAILURE in 2m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/bb604911a4a74340aa77e1d4c99e7073 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/858db8587f6c445396fad7d3634d366d : FAILURE in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abecd5d8929b46749edbe21152f854c2 : FAILURE in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1b100789ce1438e8c57a803674c9212 : FAILURE in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a71ecfcfb534190a4d29360e68000c5 : FAILURE in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/621406f0ea9948f4a3dc4f0cd9731b0e : FAILURE in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b95c38fcf2b459ab11eca8185f8318c : SUCCESS in 5m 22s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/91c062f39afe4782af2b31dbf68db3d9 : SUCCESS in 1h 18m 18s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/d42d1e0bf9804d1199d5aad1799dd390 : SUCCESS in 1h 17m 19s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/fc17b5b6819441c1b775c3167a261c0d : FAILURE in 31m 11s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/5784c7e208f34337b7a583e1a1402d79 : SUCCESS in 1h 19m 37s","accounts_in_message":[],"_revision_number":3},{"id":"1a93bf8f862a49f15ba4069489102f6873f8dfef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-02-23 14:45:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"65c98e39facf50b4449d6ca230d86ce88b095cd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 16:48:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/232fa4f65e0e4ff488207daaf40aec1d : SUCCESS in 2m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/b1486e46915e49469f5e11a07a6c8e4f : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/311611ef25ba4877accfd800d85019bd : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e976b367dec472ebc2f872b25e4e07e : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d95774414ff4777bd8f1a74bc36c8b1 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa0a1ab44e6945b5b7bc7bbabb4dc2bf : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48507ea628b04b53a6a85023f2f72a46 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8911b029c48f47b8b9b6b43f0629f020 : SUCCESS in 6m 22s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/5f6ab1a4eef4456d97c9dd399a4d737e : SUCCESS in 1h 30m 52s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/5d6d7fe816d442c78d05a09108ad38e5 : SUCCESS in 1h 31m 02s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/9bd0e47405314c59836e4279c5dee115 : SUCCESS in 1h 55m 39s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/84b78902f5c64e349f12ca2abbf373f5 : SUCCESS in 1h 29m 01s","accounts_in_message":[],"_revision_number":4},{"id":"8a20b88aad55f2476964983ea3dad9d9f2f6931b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-02-24 10:22:07.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9b5ecb27a2043a0514a2dee023f062a1a0e6e17a","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-02-28 13:13:42.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ee42d98bcf8e3abbf7f598d516a668c9dc131267","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 13:14:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ea83402bef50df2e90e511583e0a8c70bc8efbe8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 14:41:57.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f880dff1949b4a1a860726d2b2dca5d3 : SUCCESS in 2m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d88b9201b7b5463b8b40996597abc9d7 : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da4a8683436f456a809430fc5fdfbffe : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bda57f276af04430b97b5e777968b9a7 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/096b6dfab8794049a50cfc47ef3f9a71 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16dfac28d0274d1988ed4b9012315050 : SUCCESS in 6m 01s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/01973fd895a447a6805f47eb1c2608c4 : SUCCESS in 1h 22m 15s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/7a88fa98b7914fe095eeaa368c6e79d9 : SUCCESS in 1h 18m 25s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/ad3f5d0f67cd42e58046a667f88a9f86 : SUCCESS in 1h 16m 12s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/c742d6a7e10944b1bf8a6d8a51e37363 : SUCCESS in 1h 15m 27s","accounts_in_message":[],"_revision_number":4},{"id":"1b4d5e8d798b0bfb53b324ea98f2dcf10a6dbd88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 14:41:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"c8dea909f999dbc8ab079abc6c1510daf23c72de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-28 14:43:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffd12e8520bc4edd852e70f71cdc36de : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"22a944bd31e546350b94d451dc3386ba75fdab0e","revisions":{"ebe8b4cabd84d7bc131741b203ae6d4dbfa34f36":{"kind":"REWORK","_number":1,"created":"2021-02-06 12:36:15.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/54/774354/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/54/774354/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/1"}}},"commit":{"parents":[{"commit":"ac265ec815faa7db5b38d696ab5553ce21a6dec5","subject":"Import common keystone configuration from ironic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/ac265ec815faa7db5b38d696ab5553ce21a6dec5"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-02-06 11:34:24.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-02-06 12:35:38.000000000","tz":60},"subject":"[WIP] import json_rpc from ironic","message":"[WIP] import json_rpc from ironic\n\nChange-Id: I0e790cebcd086d26bf27b909f73842c3f46bbd77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/ebe8b4cabd84d7bc131741b203ae6d4dbfa34f36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/ebe8b4cabd84d7bc131741b203ae6d4dbfa34f36"}]},"branch":"refs/heads/master"},"e6ffb45eb3111f2828cb7cf19f051abf3f152c26":{"kind":"REWORK","_number":2,"created":"2021-02-06 12:51:22.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/54/774354/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/54/774354/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/2"}}},"commit":{"parents":[{"commit":"d0b70f69d34a4cc95dae5db31a258239ac5a543d","subject":"keystone: use os_service_type for service_type defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/d0b70f69d34a4cc95dae5db31a258239ac5a543d"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-02-06 11:34:24.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-02-06 12:50:42.000000000","tz":60},"subject":"[WIP] import json_rpc from ironic","message":"[WIP] import json_rpc from ironic\n\nChange-Id: I0e790cebcd086d26bf27b909f73842c3f46bbd77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/e6ffb45eb3111f2828cb7cf19f051abf3f152c26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/e6ffb45eb3111f2828cb7cf19f051abf3f152c26"}]},"branch":"refs/heads/master"},"8321db232434b74974a334f823642078de19500f":{"kind":"REWORK","_number":3,"created":"2021-02-08 12:43:23.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/54/774354/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/54/774354/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/3"}}},"commit":{"parents":[{"commit":"f4ba0a71b30f5fd5a816025977a6e4c162e55be7","subject":"keystone: use os_service_type for service_type defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/f4ba0a71b30f5fd5a816025977a6e4c162e55be7"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-02-06 11:34:24.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-02-08 12:39:27.000000000","tz":60},"subject":"[WIP] import json_rpc from ironic","message":"[WIP] import json_rpc from ironic\n\nChange-Id: I0e790cebcd086d26bf27b909f73842c3f46bbd77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/8321db232434b74974a334f823642078de19500f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/8321db232434b74974a334f823642078de19500f"}]},"branch":"refs/heads/master"},"22a944bd31e546350b94d451dc3386ba75fdab0e":{"kind":"REWORK","_number":4,"created":"2021-02-23 14:45:55.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/54/774354/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/54/774354/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/54/774354/4"}}},"commit":{"parents":[{"commit":"0b955a2bf7f69e272e7a49ef815deef7ea269c56","subject":"Merge \"Add a helper for getting partition path\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/0b955a2bf7f69e272e7a49ef815deef7ea269c56"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-02-06 11:34:24.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-02-23 14:44:25.000000000","tz":60},"subject":"Import json_rpc from ironic","message":"Import json_rpc from ironic\n\nTo be able to use the same JSON RPC implementation for ironic-inspector,\nit is now imported in ironic-lib. Some additional abstraction laters are\nadded to avoid hard dependency on oslo.context and oslo.messaging.\n\nChange-Id: I0e790cebcd086d26bf27b909f73842c3f46bbd77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/22a944bd31e546350b94d451dc3386ba75fdab0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/22a944bd31e546350b94d451dc3386ba75fdab0e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
