)]}'
{"id":"openstack%2Fironic-lib~804195","triplet_id":"openstack%2Fironic-lib~master~Id50af71faec84bd778422ed86a3b99797158afdb","project":"openstack/ironic-lib","branch":"master","topic":"automated_config_validation","hashtags":[],"change_id":"Id50af71faec84bd778422ed86a3b99797158afdb","subject":"json_rpc: setup.cfg is missing entrypoint for json_rpc\u0027s list_opts","status":"MERGED","created":"2021-08-11 04:32:56.000000000","updated":"2021-08-19 11:39:56.000000000","submitted":"2021-08-19 11:37:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804195-automated_config_validation","meta_rev_id":"7c42244d31162f72c1307a273bda9199b2cbab4d","_number":804195,"virtual_id_number":804195,"owner":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-19 11:37:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-08-19 08:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":2,"date":"2021-08-19 08:49:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-08-19 08:47:36.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":1,"date":"2021-08-19 08:49:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"}],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-11 04:34:20.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":11491,"name":"Gael Chamoulaud","email":"gchamoul@redhat.com","username":"gchamoul"},"state":"REVIEWER"},{"updated":"2021-08-11 04:34:20.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-08-11 04:34:20.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":17216,"name":"Martin Schuppert","email":"mschuppert@redhat.com","username":"mcschupp"},"state":"REVIEWER"},{"updated":"2021-08-11 04:34:20.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":20733,"name":"Rajesh Tailor","email":"ratailor@redhat.com","username":"rajesht"},"state":"REVIEWER"},{"updated":"2021-08-11 04:41:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-17 14:58:56.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-08-19 08:47:26.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"132b9d0772ce92d17428343bcfdf4b2d3a7d6b6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-08-11 04:32:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6eea74125dae45d052adc879d6d1f85ee88df93b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-08-11 04:35:46.000000000","message":"Topic set to oslo-config-validator","accounts_in_message":[],"_revision_number":1},{"id":"40f0d8a559785a5f9c5689d1f0c53d115a340654","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-08-11 04:36:35.000000000","message":"Topic oslo-config-validator removed","accounts_in_message":[],"_revision_number":1},{"id":"1fe58c2435b74257641f08e0f10acd7c1956226a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-08-11 04:36:37.000000000","message":"Topic set to automated_config_validation","accounts_in_message":[],"_revision_number":1},{"id":"c10c674c54f5ae36152c8f17a1d677918d9befbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 04:41:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d0f151bd5d27446ca619d015b4991e58 : SUCCESS in 2m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a6605cf0d3242ef8788098a024a2bf5 : SUCCESS in 4m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/126819d9c9e34d1b9cb58e8ec8100c78 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef71a4034ed4fc9b45886ad15595afb : SUCCESS in 3m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd2915dfecd144d796457672b3a3eba4 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01649c544a6d455ab1da2043aad3cd11 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/235c2abaa91b44299a036006651ec41f : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d6c5709c80a4460ae8fd901efe0e786 : SUCCESS in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f22cb7160863d9535ac817fd748c563778e66194","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-08-17 14:58:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ccfa0fe91e396820ed6fcccf8286e97aeb8dad4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-08-18 00:43:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"91e4d1ec15ee199a40ebb5ffe85b076b7531f0fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-08-18 00:43:27.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"addd8a936b86e185cf0cc5c320798b661dc67e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 00:54:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/42aabfdb44b049f1a50737d673396a92 : SUCCESS in 2m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34d2c9e6933844fabebf9f19a06deea9 : SUCCESS in 4m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b08e80ed96e747019fd9c8560ef4ea0e : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ac28ce71cdd4946834a9597d4e6635e : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7892636f45b457586d0f72efbe61f02 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c93c800c29364cfea6142c618537972f : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39a0880833e34dffb2e6c6980d2b730b : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3c02b293da748b788cd39b3c2972123 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":3},{"id":"9f00fcac03856562e78ffda18f3ab3ffa66de999","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-08-19 08:47:26.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b308cac1311a46474d1c18b3ac4fedf321c18f64","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-08-19 08:47:36.000000000","message":"Patch Set 3: -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"c50aed79a0b2bb251ccb49a7dbbb18180bb262a2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-08-19 08:49:49.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"58b7f3a816e595a7e1d58551a3f4c64342aecf50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 10:10:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"be8b29a0b7f379c25a7bf5086f056380f40333ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 11:37:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e716ca1533d445c086934d728527e068 : SUCCESS in 2m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/50b9ab2ee80844578a4531d433cbf8b8 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc9ec55c4dd4f6f9f9469859bf721eb : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d09e3da41abf4f7b807ab25532281bef : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3eb6d81a9cd7440aadbf0cb73173544b : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a988674664f1460b964b51a18e9de41b : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":3},{"id":"583a1bf4304da0b3c23cce31a3e880bae3548c63","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 11:37:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7c42244d31162f72c1307a273bda9199b2cbab4d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 11:39:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3418fc6d2b0546f88d4a88ae53fb0248 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"12bca21c0cacc2b2d422b8ae38430294f94be7e3","revisions":{"cd19b333485647a32bb19c908fed50d6d47a1c2a":{"kind":"REWORK","_number":1,"created":"2021-08-11 04:32:56.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/95/804195/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/95/804195/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/95/804195/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/95/804195/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/95/804195/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/95/804195/1"}}},"commit":{"parents":[{"commit":"eefe38a3ef440b9a469e0690dc073e7a443b68ba","subject":"json_rpc: list_opts() should return tuple","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/eefe38a3ef440b9a469e0690dc073e7a443b68ba"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-08-11 04:31:45.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-08-11 04:31:45.000000000","tz":-240},"subject":"json_rpc: setup.cfg is missing entrypoint for json_rpc\u0027s list_opts","message":"json_rpc: setup.cfg is missing entrypoint for json_rpc\u0027s list_opts\n\noslo.config is relying on the list_opts entrypoint to automatically\nintegrate the json_rpc options.\n\nChange-Id: Id50af71faec84bd778422ed86a3b99797158afdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cd19b333485647a32bb19c908fed50d6d47a1c2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cd19b333485647a32bb19c908fed50d6d47a1c2a"}]},"branch":"refs/heads/master"},"5282a40f97bf134707cf277b8d2a08b83561b137":{"kind":"NO_CHANGE","_number":2,"created":"2021-08-18 00:43:11.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/95/804195/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/95/804195/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/95/804195/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/95/804195/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/95/804195/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/95/804195/2"}}},"commit":{"parents":[{"commit":"f6ddd09d50d3181a7cc2f9f4e201c26522ee8143","subject":"json_rpc: list_opts() should return tuple","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/f6ddd09d50d3181a7cc2f9f4e201c26522ee8143"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-08-11 04:31:45.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-08-18 00:43:11.000000000","tz":0},"subject":"json_rpc: setup.cfg is missing entrypoint for json_rpc\u0027s list_opts","message":"json_rpc: setup.cfg is missing entrypoint for json_rpc\u0027s list_opts\n\noslo.config is relying on the list_opts entrypoint to automatically\nintegrate the json_rpc options.\n\nChange-Id: Id50af71faec84bd778422ed86a3b99797158afdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/5282a40f97bf134707cf277b8d2a08b83561b137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/5282a40f97bf134707cf277b8d2a08b83561b137"}]},"branch":"refs/heads/master","description":"Rebase"},"12bca21c0cacc2b2d422b8ae38430294f94be7e3":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-08-18 00:43:27.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/95/804195/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/95/804195/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/95/804195/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/95/804195/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/95/804195/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/95/804195/3"}}},"commit":{"parents":[{"commit":"f6ddd09d50d3181a7cc2f9f4e201c26522ee8143","subject":"json_rpc: list_opts() should return tuple","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/f6ddd09d50d3181a7cc2f9f4e201c26522ee8143"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-08-11 04:31:45.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-08-18 00:43:27.000000000","tz":0},"subject":"json_rpc: setup.cfg is missing entrypoint for json_rpc\u0027s list_opts","message":"json_rpc: setup.cfg is missing entrypoint for json_rpc\u0027s list_opts\n\noslo.config is relying on the list_opts entrypoint to automatically\nintegrate the json_rpc options.\n\nStory: 2009126\nTask: 43038\n\nChange-Id: Id50af71faec84bd778422ed86a3b99797158afdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/12bca21c0cacc2b2d422b8ae38430294f94be7e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/12bca21c0cacc2b2d422b8ae38430294f94be7e3"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
