)]}'
{"id":"openstack%2Fironic-lib~813570","triplet_id":"openstack%2Fironic-lib~stable%2Fwallaby~Ib4b9c270709f5994bc2c031c339822d7d6b9356d","project":"openstack/ironic-lib","branch":"stable/wallaby","topic":"automated_config_validation-stable/wallaby","attention_set":{},"removed_from_attention_set":{"27419":{"account":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"last_update":"2021-10-15 00:47:22.000000000","reason":"removed on reply"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2021-10-13 07:25:50.000000000","reason":"Riccardo Pittau replied on the change"},"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2021-10-15 04:22:08.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2021-10-13 09:05:43.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ib4b9c270709f5994bc2c031c339822d7d6b9356d","subject":"json_rpc: list_opts() should return tuple","status":"MERGED","created":"2021-10-12 02:42:52.000000000","updated":"2021-10-15 04:24:45.000000000","submitted":"2021-10-15 04:22:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"813570-automated_config_validation-stable/wallaby","cherry_pick_of_change":804194,"cherry_pick_of_patch_set":2,"meta_rev_id":"7155225e6e703411742f47d746fb0e17ba16e564","_number":813570,"virtual_id_number":813570,"owner":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-15 04:22:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-10-13 07:25:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-10-13 09:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-13 09:05:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-12 04:16:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-12 11:23:45.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-10-12 11:23:45.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-10-12 11:23:45.000000000","updated_by":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"feb6541124d8f7b163624e1807afe40a7889f5d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-10-12 02:42:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c6601d4773ba62a1a9ec14d18bc8d4ef25baed58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-12 04:16:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23ffa3c1abe24db9912843a58c1e3155 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c58f31218e3243468051894f093aaa3d : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bb30ae42c064e0a98bef30901432c89 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e46049e88ef4b1bba089e07320b0c55 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/952bb3643dad4be5a568049c1637aed5 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9963f7c0bb14bbea779e22d2803ea9c : SUCCESS in 7m 10s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/a0973dd938a94a349d9d5694c5a93a38 : SUCCESS in 1h 29m 45s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/4187665c02a142c5a2a284f3c547e35e : SUCCESS in 1h 31m 46s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/302da43b01f845e399e4aa6360343eae : SUCCESS in 1h 29m 36s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/3d9db66ee8cb40af9f4718fd6d0aaaf4 : SUCCESS in 1h 26m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b333803d1c7f501a3034cfead9ca5b8d0b1cd94c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-10-13 07:25:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"786426a75e9dbffa6002c62c3dff4935084f38d7","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-10-13 09:05:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c6a2953f7ea4be0f40bd23387d0d911bb087a237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 09:06:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef0cb5feb33ba93d73ebca50b5387cbf430a6d57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 10:51:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2f496859b2b4c8aa4df09ac657d05fb : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1a9e9ced89746c49e6c751143327f27 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1b3f9e394a643ef954c9b287d2cffa2 : POST_FAILURE in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/252906a88a534f45a8c9ad8ae3bc8c1c : SUCCESS in 6m 07s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/a3ad22820e77448699096e19564f543e : FAILURE in 53m 17s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/255cd974986648dd9f9f81dba9d0b230 : SUCCESS in 1h 43m 23s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/77ba6a10b28f409d8b169417e340f549 : SUCCESS in 1h 19m 10s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/cf3dbf7417364d1694ed4ee7f8574fec : SUCCESS in 1h 31m 00s","accounts_in_message":[],"_revision_number":1},{"id":"a38cfb3e7c12ed77bfe2b56978c0786576c86a1e","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-10-13 13:25:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f31a665b1126c19dbcd6deecb148c16cbdcf156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-13 15:01:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2997c12f922747e6b8e5aa9211e812ab : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b2ac732dce04b8fa21e5e8d7a26f465 : SUCCESS in 7m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6b525b92c274b09859c05dc026879a7 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dee237fc798b4f928c0d29fbe3bb7441 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5515ef9351764ad9aca05a4881d3a830 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f0c16bd49eb4a39894fa30f2c6b15b0 : SUCCESS in 8m 18s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/e77296bba5854417b4479743eb065794 : SUCCESS in 1h 33m 00s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/0d9442f4e39e4d3d8ee9e185135a537d : FAILURE in 1h 22m 04s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/163ab96dda184d5aaaf9426e87cb78aa : SUCCESS in 1h 17m 52s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/c5fe2df9403c4a408a62637a678cd2f1 : SUCCESS in 1h 29m 15s","accounts_in_message":[],"_revision_number":1},{"id":"968d335b9ff86f028a426859e67159cea9f82305","tag":"compute.df.review","author":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"date":"2021-10-15 00:47:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"349587c37e827487962d039c5183fa4b58ff6c14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 02:25:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0fc3e9aeee64be69a06b9424357a7a2 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f93efb93c1f4f4f9efe6ee1c32378a5 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa5603010fc04cce9c6efeaf4c560637 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea48507bde2d43cf811103702d25cc2c : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/511ba177f8124a23af04bb4ef1c39002 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62db8af9ef1846c1a13982ca12f1277b : SUCCESS in 6m 22s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/3ae92d74cbc944e68bbcfbba535d3ece : SUCCESS in 1h 29m 59s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/95a13766e325452795968c59b7df2546 : SUCCESS in 1h 25m 48s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/fd06b264e73848ccb651b63adbbddd3d : SUCCESS in 1h 31m 47s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/b537528dafdb49a194cd7d9138b85108 : SUCCESS in 1h 26m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8575c31aec0095a8884c6e7a9435acafd85d3828","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 02:35:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"742199a7d22961cab9453a8255087b04142a5ec4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 04:22:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70406f7e330e4d44bf3c170c8666d48b : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52439e42d9424c57b11caaf495d1eb8a : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c89a9beb30664693aebe9b24e3ebae50 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/828fef52f57642b485688479a4fe39e3 : SUCCESS in 9m 52s\n- ironic-lib-wholedisk-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/df684f87c6f64e05b59bf70689b2ce4c : SUCCESS in 1h 44m 29s\n- ironic-lib-partition-bios-ipmi-direct-src https://zuul.opendev.org/t/openstack/build/8353725803dd48cabf918e95a5c407e9 : SUCCESS in 1h 36m 00s\n- ironic-lib-partition-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/c5753179911c4d4f90b38afeaee8ed1c : SUCCESS in 1h 27m 45s\n- ironic-lib-wholedisk-bios-ipmi-iscsi-src https://zuul.opendev.org/t/openstack/build/0090aad89f344403874dfcef86fc43d7 : SUCCESS in 1h 17m 27s","accounts_in_message":[],"_revision_number":1},{"id":"aece6abe02a315141700c209b68fc0dfe63b6d88","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 04:22:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7155225e6e703411742f47d746fb0e17ba16e564","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 04:24:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b363e777d2e4d98bc365f07ac2f29e7 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15063ca2224e85d1a963190f896318b89640ca40","revisions":{"15063ca2224e85d1a963190f896318b89640ca40":{"kind":"REWORK","_number":1,"created":"2021-10-12 02:42:52.000000000","uploader":{"_account_id":27419,"name":"David Vallee Delisle","email":"me@dvd.dev","username":"dvd"},"ref":"refs/changes/70/813570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/70/813570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/70/813570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/70/813570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/70/813570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/70/813570/1"}}},"commit":{"parents":[{"commit":"daf31553894e84a015fc0aa8d97e203dd2d7d678","subject":"Avoid using blkid in favour of lsblk and parted","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/daf31553894e84a015fc0aa8d97e203dd2d7d678"}]}],"author":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-08-11 04:29:35.000000000","tz":-240},"committer":{"name":"David Vallee Delisle","email":"dvd@redhat.com","date":"2021-10-12 02:42:52.000000000","tz":0},"subject":"json_rpc: list_opts() should return tuple","message":"json_rpc: list_opts() should return tuple\n\nlist_opts() should return a tuple otherwise\noslo-config-(genera|valida)tors are unable to parse its output.\n\nStory: 2009126\nTask: 43039\n\nChange-Id: Ib4b9c270709f5994bc2c031c339822d7d6b9356d\n(cherry picked from commit f6ddd09d50d3181a7cc2f9f4e201c26522ee8143)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/15063ca2224e85d1a963190f896318b89640ca40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/15063ca2224e85d1a963190f896318b89640ca40"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
