)]}'
{"id":"openstack%2Fironic-lib~830319","triplet_id":"openstack%2Fironic-lib~master~I79a4b189a7fb220a97d83601002339ebfae45104","project":"openstack/ironic-lib","branch":"master","topic":"bindep-py3-only","attention_set":{},"removed_from_attention_set":{"24245":{"account":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"last_update":"2022-07-05 12:26:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I79a4b189a7fb220a97d83601002339ebfae45104","subject":"Remove python version 2 pkgs from bindep","status":"MERGED","created":"2022-02-22 07:50:20.000000000","updated":"2022-07-05 12:29:01.000000000","submitted":"2022-07-05 12:26:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830319-bindep-py3-only","meta_rev_id":"97d6c57a7d94d9fbc7dee0ba4f6dfa8bb7d47dd4","_number":830319,"virtual_id_number":830319,"owner":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 12:26:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-06-27 11:04:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-06-28 08:33:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":2,"date":"2022-07-05 10:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":1,"date":"2022-07-05 10:52:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 09:31:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 09:49:38.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-03-03 10:48:12.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2022-06-28 08:33:42.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"197aa3b2f322e4d6d9b057b950ba3fae4bcdf869","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-22 07:50:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b35f332241e4e77a74621cb77dad9bac95d5df52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 09:31:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2dc1e724f3348dfa0ae6a18fd9bc4dc : SUCCESS in 4m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3dd0f19c6ba442e083ef41fc0209b283 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b423ee2d0f084d6282054a454307cf3c : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f2d397472f24483b82b8daaa095209a : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9c1821daa6546ea8dd39558a2bfd28e : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd9d1d8d76ab43d781cb86c0d2af6420 : SUCCESS in 6m 08s\n- ironic-lib-wholedisk-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/2bdea27180734dd3974414aeae6d1051 : SUCCESS in 1h 27m 48s\n- ironic-lib-partition-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/ec1b3cd1943546adbe9d7bfea6ab88eb : FAILURE in 1h 36m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e83be48f06ba9558211b0b3447978f3b2fbda5c6","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-02-22 09:49:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf0168180ce885a71bb81cff73d200f9453522dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 11:38:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4206e1e2a7e946bc8f4f2e6442c66d70 : SUCCESS in 4m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26b714fbf2384fafbc16ce08cebaced1 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4725c0f9cb84773b020d0633cffe18b : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b63ec057c0144643b18e464ea58a21fa : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52147bb4a28a4e82ab062dc262b34ab9 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b64ce769197b4cf5a3e7edec5aaa90c7 : SUCCESS in 6m 27s\n- ironic-lib-wholedisk-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/b7261a39a37f4670b066a81135321f89 : FAILURE in 1h 46m 28s\n- ironic-lib-partition-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/37806ed2877548319b305513345c0525 : FAILURE in 1h 25m 14s","accounts_in_message":[],"_revision_number":1},{"id":"9b87cae99a2a778c882fb207d1a652cdbfede115","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-22 12:44:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a3f2f0e270c916c90185590b8317656f95425a24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 14:49:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edc45c8e8c444528b1859d464815dc9a : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0129eee2dba44c2a559d1fc29178731 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f54cd88a0e4e90a630574cd6267795 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c3c0608ae3e4f5486af072f5b75d667 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/300e4e9c797b435593ef49f3aa217101 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5032579b0dd645fab1182e6ffa5f14f0 : SUCCESS in 7m 16s\n- ironic-lib-wholedisk-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/83c86e9520c54d8e80a580862fceaaab : SUCCESS in 1h 58m 12s\n- ironic-lib-partition-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/03786ee42ea6455f93381ac1b51642a3 : FAILURE in 2h 01m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ad9ebb42f61bdf8305a41b0bf4f6221ed89113a5","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-02-28 22:03:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5041c2b58587b1e4458964c829c87b981d07b99","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 23:51:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c748282936044f828b032e3308530f92 : SUCCESS in 5m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d054b41486084c0aad1e46c3d951cd26 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f91941528bea4b8b942e48ab1b6d366e : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f16c587e607f411eafc0f311024b584f : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff2993b422be4014a33daf32fa83d1c2 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d42f1056230475ebbcfbe96beac4310 : SUCCESS in 6m 57s\n- ironic-lib-wholedisk-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/00c2ff5699624ec48a3dd7fe74a4c7eb : SUCCESS in 1h 42m 29s\n- ironic-lib-partition-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/06e7556c777942838ad96402af4ba7a0 : FAILURE in 1h 38m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e1e2fd830f443e5a42ff8f7f371eea2a9e239f79","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-03-01 15:39:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a1ceb1f93acf30f810c588c3180d1a5b2800e57a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 17:45:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/406d299b74c64787a0288fc5f24ea3c9 : SUCCESS in 6m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b35e1cdd8b0f4f6a8e5faca0cd9d06a5 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/720f3826dd394dd78d1db3305e5ea30b : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f01ce28833504fccaf657cda9e90f451 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b33a06dd19f4ef99942e194c66dbb72 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5606a880a5064d16933776d28d835ade : SUCCESS in 8m 44s\n- ironic-lib-wholedisk-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/25727e169bda497d889e834653c33a89 : SUCCESS in 1h 45m 05s\n- ironic-lib-partition-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/975257cd8a374b65a98e87196d440463 : FAILURE in 2h 01m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e8445f62cfea975f3b80d72b288c1eed1c9b7e8f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-03-03 10:48:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93087290f5fbca5cff0bff59c66f9dfad12a6109","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-03-04 14:57:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"87c874662bf29e0ba04a18e55dc3a9ed1ad8ed4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-04 16:42:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b622ef7d1b4f4747891457780c93ec3d : SUCCESS in 5m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb43ef8a69a54af89de9519ee288a015 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a8e09ae0024e27b943e2bc2ec886b5 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36aba7d5c23b4e998496d427a54ebf67 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b013e065a7264ca2825bf2cb6f2e53ac : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2cddea62e4343bdbe6a81d47e35137d : SUCCESS in 6m 58s\n- ironic-lib-wholedisk-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/ec9b208fcf064fd293b18051a7b18e6c : SUCCESS in 1h 30m 34s\n- ironic-lib-partition-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/495445c934614353825292ece79ab206 : FAILURE in 1h 40m 29s","accounts_in_message":[],"_revision_number":2},{"id":"dfa48dd3f5f1742ceca9f49760d7839189c5f4df","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-03-26 02:22:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fa351fe4d32065578cfc6a98db8445b1187fe55e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-26 04:22:51.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb49406516b346ec97ccf973688ffb7a : SUCCESS in 6m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/43ca493ce472489b9a03bf3419e46aae : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2235e411f3864b35a87fa716d687346f : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15a4b873d8f54f5b804137addc05e607 : SUCCESS in 7m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/314606fa40684918ba441091f5a7a849 : SUCCESS in 8m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea1f59fcbd4540c38bf668072db542c2 : SUCCESS in 10m 28s\n- ironic-lib-wholedisk-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/9ebb2eb089854316906266f43d99c317 : FAILURE in 1h 38m 17s\n- ironic-lib-partition-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/c7540a0b7cb8401aa6a9eff1c8c7fd81 : FAILURE in 1h 54m 05s","accounts_in_message":[],"_revision_number":2},{"id":"3ed03864e7b4656a2988c4f917f40a34e2a90a02","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-03-29 07:15:13.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7b2d2a50caa2d09c47c8e19258dac1263e550c79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-29 07:49:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7c4bc67441d4714b2218741840111d3 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8db55d253fb941c4ac9d2241639d16fb : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872f7eaf01c648f687f68ed53e6d7d76 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95ad32cd5abf42ccbe9d43c73f5e3bf2 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e07cd512b344802b2abf1adbab54bab : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13154b38c4214e868829d204b3ea29e8 : SUCCESS in 8m 12s\n- ironic-lib-wholedisk-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/e077e1ee7c0c48349c6980e741d2d930 : FAILURE in 32m 59s\n- ironic-lib-partition-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/75268d4d1e1340c692878f83e814b94f : FAILURE in 31m 49s","accounts_in_message":[],"_revision_number":2},{"id":"2b5ec9f4c0f309badff578b7f26b682a3efa63e7","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-04-05 09:35:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"27a113ff69ffdf089463e8641cd74f7a1d4979b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-05 11:31:13.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98178601959c4e9cb59e40dd26ea6fa7 : SUCCESS in 4m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/671e9c9c226643c9b97e567535ddeea2 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79f07e660c664cad94856feb036e064e : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54a6c872effd4a768ad8a435265d7c9d : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fde96813270746ebb4cb5a3a95ff1ac2 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66051777c86246bda109a4c4f8136cbc : SUCCESS in 7m 31s\n- ironic-lib-wholedisk-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/6f5bbdf24467452e8db0c82d7e4a4b32 : SUCCESS in 1h 39m 13s\n- ironic-lib-partition-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/f62471974cca4ade9b5ed99fbb7f02de : FAILURE in 1h 47m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a93f0a42e1a6f1f44b3fcdfa5cacace230f62699","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-06-14 13:58:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1269e2ce69f1d877dbe855b159d554a516e4b6f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 15:36:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84cf6e53235f4b659a8abcd4dbea7811 : SUCCESS in 3m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35eb6f9ac5cb46478161140cb210530a : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd1b86932cde4eca89223e9d505af33b : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fd5387fbff443d3966471161f1a81ee : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d40f9d4d9a19433b897e2e1ce1a84ac1 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/52346e75a20c4ccca188acff723994a5 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08e8b629262242e69b339af4c3f95423 : SUCCESS in 6m 06s\n- ironic-lib-local-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/2fccc585180343249a078c609575814d : SUCCESS in 1h 32m 44s\n- ironic-lib-netboot-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/aa4812f6fe154c728cdcc8de4ce697f2 : FAILURE in 34m 43s","accounts_in_message":[],"_revision_number":2},{"id":"b0b0cbfc62cf8f6d5bf3073c3344dbead126373d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"date":"2022-06-27 11:04:45.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fd8bc127cd0ea03ba308f2b73965fa0b2c74e018","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 12:50:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2cd598e53e24c738ef37d8fa7a7dec4 : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03fc700ed01e478eab0e924cb62b6169 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3d9bc5847d847b48d5e1158303b962d : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/368deb423fa142fcb8390c75e353dd99 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba55a11f7d4c457b9f1b029d705ff796 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc3ca0c4e8c14ee18781297b0ae31663 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcd0e009ce6746548db6512c11234ec1 : SUCCESS in 6m 25s\n- ironic-lib-local-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/edd9b0d93c74449d84a47cbd019e6782 : SUCCESS in 1h 39m 54s\n- ironic-lib-netboot-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/79454b3bcf88480e934829ec78e7a348 : SUCCESS in 1h 18m 00s","accounts_in_message":[],"_revision_number":3},{"id":"61247b1f2b9a1daf719188a64d99b457b066a923","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2022-06-28 08:33:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"04393dae66ea5e566d112b908f098edb058cc0a9","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-07-05 10:52:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b9f973a8990e798d3da6529a2704b956691b4a9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 10:53:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5ba89db392ef5ae07096128dfcf77aa4aaaa94bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 12:26:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f64bd126c984be197bf27e68a17eef9\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecfd460d08f340a28553edca5bcfc13d : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61caed2fcbee43e6b6de971e3db32cf0 : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1269cfc275bb4239aa2fa4cbbeaf0a8d : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ef99fafe9c54fc09ac7069e79eecc0d : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b338f729ac45ee81b69f72046581bc : SUCCESS in 4m 57s\n- ironic-lib-local-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/2a05d75d22c1445394b38291a6b3147f : SUCCESS in 1h 29m 16s\n- ironic-lib-netboot-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/5689e68ee9ec4a5498689433cc521f6a : SUCCESS in 1h 26m 25s","accounts_in_message":[],"_revision_number":3},{"id":"2e577831cfc71b2d6eb4b13b9bef19697ab2ccf4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 12:26:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"97d6c57a7d94d9fbc7dee0ba4f6dfa8bb7d47dd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 12:29:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e0c415612be414ebded47937debcf55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e75af8beb8848cea87bd1b84d7c7dae : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0aa35a0edba4cce619d35c9964e975eadb79f9f8","revisions":{"cb108cd69b9d8bf53d9b89cd784dafb94425bf43":{"kind":"REWORK","_number":1,"created":"2022-02-22 07:50:20.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/19/830319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/19/830319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/19/830319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/19/830319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/19/830319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/19/830319/1"}}},"commit":{"parents":[{"commit":"e205816df70c8acbecae2be9484699f3eb832e1b","subject":"Merge \"Add platform:base-py[2|3] to bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/e205816df70c8acbecae2be9484699f3eb832e1b"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-02-22 07:48:54.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-02-22 07:48:54.000000000","tz":60},"subject":"Remove python version 2 pkgs from bindep","message":"Remove python version 2 pkgs from bindep\n\nFollow up to change: Iee5ac3a778f8106b7f2e96cd9294e2da28fe4def\n\nSince Python version 2 is not supported we can remove these.\n\nChange-Id: I79a4b189a7fb220a97d83601002339ebfae45104\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cb108cd69b9d8bf53d9b89cd784dafb94425bf43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cb108cd69b9d8bf53d9b89cd784dafb94425bf43"}]},"branch":"refs/heads/master"},"a3fb6f0c22d5270ef1e4870b2247881abb1c9c39":{"kind":"REWORK","_number":2,"created":"2022-03-04 14:57:00.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/19/830319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/19/830319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/19/830319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/19/830319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/19/830319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/19/830319/2"}}},"commit":{"parents":[{"commit":"e205816df70c8acbecae2be9484699f3eb832e1b","subject":"Merge \"Add platform:base-py[2|3] to bindep\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/e205816df70c8acbecae2be9484699f3eb832e1b"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-02-22 07:48:54.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-03-04 14:56:42.000000000","tz":0},"subject":"Remove python version 2 pkgs from bindep","message":"Remove python version 2 pkgs from bindep\n\nFollow up to change: Iee5ac3a778f8106b7f2e96cd9294e2da28fe4def\n\nSince Python version 2 is not supported we can remove these.\n\nChange-Id: I79a4b189a7fb220a97d83601002339ebfae45104\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/a3fb6f0c22d5270ef1e4870b2247881abb1c9c39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/a3fb6f0c22d5270ef1e4870b2247881abb1c9c39"}]},"branch":"refs/heads/master"},"0aa35a0edba4cce619d35c9964e975eadb79f9f8":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-06-27 11:04:45.000000000","uploader":{"_account_id":24245,"name":"Harald Jensås","email":"hjensas@redhat.com","username":"harald.jensas"},"ref":"refs/changes/19/830319/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/19/830319/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/19/830319/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/19/830319/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/19/830319/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/19/830319/3"}}},"commit":{"parents":[{"commit":"5662b92fda78e123dd3adfcc2c818ff251af6ec1","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/5662b92fda78e123dd3adfcc2c818ff251af6ec1"}]}],"author":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-02-22 07:48:54.000000000","tz":60},"committer":{"name":"Harald Jensås","email":"hjensas@redhat.com","date":"2022-06-27 11:04:45.000000000","tz":0},"subject":"Remove python version 2 pkgs from bindep","message":"Remove python version 2 pkgs from bindep\n\nFollow up to change: Iee5ac3a778f8106b7f2e96cd9294e2da28fe4def\n\nSince Python version 2 is not supported we can remove these.\n\nChange-Id: I79a4b189a7fb220a97d83601002339ebfae45104\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/0aa35a0edba4cce619d35c9964e975eadb79f9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/0aa35a0edba4cce619d35c9964e975eadb79f9f8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
