)]}'
{"id":"openstack%2Fironic-lib~855728","triplet_id":"openstack%2Fironic-lib~master~Ie049a8c506408de45bd3fde13a957945edeb9f6f","project":"openstack/ironic-lib","branch":"master","topic":"release-health-check","hashtags":[],"change_id":"Ie049a8c506408de45bd3fde13a957945edeb9f6f","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2022-09-03 20:28:35.000000000","updated":"2022-09-07 09:17:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5fc255f5d9542bfc9a26a9b068ad80d1eafc400a","_number":855728,"virtual_id_number":855728,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-03 22:24:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-03 22:24:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6bbe1bf58237dfb6d86120909432251027e07738","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-03 20:28:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"423ef4cec28318af1d61aa57db553781af993271","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 22:24:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86d4af1f47eb46ce9aea06c27b32d021\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79532749d8654d62a1f907d312d0f872 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c7199ed2a7b4cec95e931de8be86ec7 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33400e33fdaf486da60dfe402c3e4866 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecbfdb613c9c4a7fb433b1e7e1619997 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aabfd904e9404c26baefb08679180d99 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/667e275ad22e4033aa9943b9cc369f48 : SUCCESS in 7m 21s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/f2263bdb840b4737bc4a2e405afc81a0 : SUCCESS in 1h 49m 22s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/95dfe3b84d0140409846a9e02106f9cc : SUCCESS in 1h 23m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5fc255f5d9542bfc9a26a9b068ad80d1eafc400a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-07 09:17:30.000000000","message":"Abandoned\n\nGate check passed, no further need for this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d7757ce25a601ad0dd2d1235d49259d612c8580b","revisions":{"d7757ce25a601ad0dd2d1235d49259d612c8580b":{"kind":"REWORK","_number":1,"created":"2022-09-03 20:28:35.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/28/855728/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/28/855728/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/28/855728/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/28/855728/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/28/855728/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/28/855728/1"}}},"commit":{"parents":[{"commit":"340a4b264a09c961aa5dfb7886e0e919daf1c051","subject":"Merge \"remove unicode prefix from code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/340a4b264a09c961aa5dfb7886e0e919daf1c051"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:28:23.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:28:33.000000000","tz":120},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: Ie049a8c506408de45bd3fde13a957945edeb9f6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/d7757ce25a601ad0dd2d1235d49259d612c8580b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/d7757ce25a601ad0dd2d1235d49259d612c8580b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
