)]}'
{"id":"openstack%2Fironic-lib~876143","triplet_id":"openstack%2Fironic-lib~stable%2F2023.1~I8706de87e39fb44323b47b785c31116c84299c0f","project":"openstack/ironic-lib","branch":"stable/2023.1","topic":"create-2023.1","hashtags":[],"change_id":"I8706de87e39fb44323b47b785c31116c84299c0f","subject":"Update .gitreview for stable/2023.1","status":"MERGED","created":"2023-03-02 14:13:16.000000000","updated":"2023-03-02 15:06:44.000000000","submitted":"2023-03-02 15:04:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876143-create-2023.1","meta_rev_id":"9f3fed3913c78c6367fbbbd9c2e70299017db08d","_number":876143,"virtual_id_number":876143,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-02 15:04:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2023-03-02 14:13:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2023-03-02 14:13:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-02 14:13:35.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2023-03-02 14:30:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"20b3cd47ce09419fda3047f0893c0552a72cda1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2023-03-02 14:13:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88fb893cf1b149208daad71885a17027bec2d567","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-03-02 14:13:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21262092ebe7797d1055ed775b8bb14fd775b46c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 14:30:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fc03777e9564e23a083b6c289d0bba5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0ba288085b948fb93752bc8270dc43d : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9be494b160741a7ae53018807831d8e : SUCCESS in 2m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57ef78117d44441cbe4071937fe1edc5 : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94de9baa0b3548c5a2f3f0e8972d5a86 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c48b750f9864845a2caedfe022978f2 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":1},{"id":"3f683a9f9843e8fd07657399c3c966f9c41bdaed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 14:31:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b555446f2653c4d8c946b14d14adb076740907a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:04:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff048fb34f4e4f7db83d671665538afc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f842ad6bebd845f39c76f9eadacdf8bc : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f3ab36c5b864ff48ae9beb543208324 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13adcc91ecf24c8cb826494d2bc148ec : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/449d8a2cf21b4a6cb3855d7687bcbd4e : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"aa9bcc14d3dbfff84772afa9f52ebcbc50368f25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:04:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9f3fed3913c78c6367fbbbd9c2e70299017db08d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:06:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/534e66bd512148589377f62d882df68c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b312d683486a4007b939369f90c8a473 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8ef3bfb3cf1ac88d1433fac32107a66fd626b051","revisions":{"8ef3bfb3cf1ac88d1433fac32107a66fd626b051":{"kind":"REWORK","_number":1,"created":"2023-03-02 14:13:16.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/43/876143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/43/876143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/43/876143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/43/876143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/43/876143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/43/876143/1"}}},"commit":{"parents":[{"commit":"71a765572302176caced8c0b934dc3c13c726673","subject":"Merge \"Provide an interface to store metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/71a765572302176caced8c0b934dc3c13c726673"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-03-02 14:13:15.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2023-03-02 14:13:15.000000000","tz":0},"subject":"Update .gitreview for stable/2023.1","message":"Update .gitreview for stable/2023.1\n\nChange-Id: I8706de87e39fb44323b47b785c31116c84299c0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/8ef3bfb3cf1ac88d1433fac32107a66fd626b051"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/8ef3bfb3cf1ac88d1433fac32107a66fd626b051"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[]}
