)]}'
{"id":"openstack%2Fironic-lib~911871","triplet_id":"openstack%2Fironic-lib~stable%2F2024.1~I076dbcf521ccd0e5cf64ff0c13ac4a119bdfc580","project":"openstack/ironic-lib","branch":"stable/2024.1","topic":"create-2024.1","hashtags":[],"change_id":"I076dbcf521ccd0e5cf64ff0c13ac4a119bdfc580","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","status":"MERGED","created":"2024-03-07 15:35:25.000000000","updated":"2024-03-12 22:30:39.000000000","submitted":"2024-03-12 22:29:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"911871-create-2024.1","meta_rev_id":"c58ef3871c1adafeac6c40fd7f661e4c8d87f61d","_number":911871,"virtual_id_number":911871,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-12 22:29:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2024-03-07 15:44:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2024-03-07 15:44:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-07 15:44:46.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2024-03-07 15:59:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-12 21:37:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"754534ab6a5a1527138dd30c783ee2582bdaebef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-03-07 15:35:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee97ecc8a44dded07ff069c977854b678464d03a","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-03-07 15:44:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1be11924c3e93a338c3f125600de89b66669e147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 15:59:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c01fa1358a6b4227bb729ee17394b2d1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5803061e7b641909b9cbc7b68ca72c1 : SUCCESS in 3m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42928db0f0394a228be717ec77924526 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08b032f9a08c483f8b68139046b1cbfe : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea88a1c5b5854c9a89ae4aa06a959f79 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8652eba9c8e2434d84d491ec56d87c7a : SUCCESS in 8m 05s\n- ironic-lib-tox-codespell https://zuul.opendev.org/t/openstack/build/f5009c2bd2b645a7bd570274f2ddf5dc : SUCCESS in 3m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ec3b6b17255b146b1ff9e16e082c5fb2a00eddb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 15:59:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fe3c86054bc61023e8c3217b8c77b8902a198b98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-07 16:07:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0807e42a7364f58bc5d17c8bf64b8c9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdaa35d061c3495bbe00c97d5a29922a : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18fb8a36053349468d538924981af3aa : CANCELED\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba45822e9091439a9a097b7ddb4c0ec6 : CANCELED\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b051bdfa2c9643b3b0cfe3cfdc131c96 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"11161daf0e7c5d0e773acfc6da15d0661ae15bba","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-12 21:37:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb4376ee2bbaf6fe5b1b0fd5d52983cb4539243e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 22:11:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65c4d24847374dd09eaa01961fabfdbc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6ab41701907405baaff1b162bae5d2d : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f659fdce12d4e688d882a3a61716a49 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db1e9acaf6b74a07b98c02f24518de6d : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d63902fa48174f7596d1912a1f2f6f9e : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/509b04b4c3f94facb04b29f1b9c6ade3 : SUCCESS in 5m 04s\n- ironic-lib-tox-codespell https://zuul.opendev.org/t/openstack/build/9ab1d7e796e94ebf9976694f58deb3bd : SUCCESS in 2m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c52b22c57296ceb5de6a2236a6026967c9f67b6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 22:11:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1d4e37b3c542c5cf24f1549fa03ed9dd62bdb33f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 22:29:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/190e9b661e084d0cbb1cc92876c132f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f3c705845774615805bcfa71662695f : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2b60d597a1748d0b03d526f7e9cd4e0 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb413aa1f7df46e2b6734a27f93b3096 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73a4122d12ec49c5ada4112d7e28d533 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":1},{"id":"92d084305e001e914399c359b28940abb4879a07","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 22:29:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c58ef3871c1adafeac6c40fd7f661e4c8d87f61d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-12 22:30:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff51889f03874b0a9d65d4832f39e8fe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa3bc2f579344a85b291ddd7286fe857 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9be44cf9fda1e8ba86c313025a66cdc9f0220e02","revisions":{"9be44cf9fda1e8ba86c313025a66cdc9f0220e02":{"kind":"REWORK","_number":1,"created":"2024-03-07 15:35:25.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/71/911871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/71/911871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/71/911871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/71/911871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/71/911871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/71/911871/1"}}},"commit":{"parents":[{"commit":"7b9c14e98599a0f6e3e058cad1209624a03e7101","subject":"Update .gitreview for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/7b9c14e98599a0f6e3e058cad1209624a03e7101"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-07 15:35:24.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-03-07 15:35:24.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","message":"Update TOX_CONSTRAINTS_FILE for stable/2024.1\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2024.1 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I076dbcf521ccd0e5cf64ff0c13ac4a119bdfc580\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/9be44cf9fda1e8ba86c313025a66cdc9f0220e02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/9be44cf9fda1e8ba86c313025a66cdc9f0220e02"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
