)]}'
{"id":"openstack%2Fironic-lib~917937","triplet_id":"openstack%2Fironic-lib~master~I9e6d1d7fb7342daa64c73244674c0bc3c420963f","project":"openstack/ironic-lib","branch":"master","topic":"4096-sector-size","attention_set":{},"removed_from_attention_set":{"36976":{"account":{"_account_id":36976,"name":"Tudor Domnescu","display_name":"Tudor Domnescu","email":"tdomnesc@redhat.com","username":"tdomnesc"},"last_update":"2024-05-03 14:50:25.000000000","reason":"\u003cGERRIT_ACCOUNT_36976\u003e replied on the change","reason_account":{"_account_id":36976,"name":"Tudor Domnescu","display_name":"Tudor Domnescu","email":"tdomnesc@redhat.com","username":"tdomnesc"}},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2024-05-03 14:46:54.000000000","reason":"removed on reply"}},"hashtags":["ironic-week-prio"],"change_id":"I9e6d1d7fb7342daa64c73244674c0bc3c420963f","subject":"destroy_disk_metadata: support 4096 sector size","status":"MERGED","created":"2024-05-02 09:34:16.000000000","updated":"2024-05-03 16:34:34.000000000","submitted":"2024-05-03 16:33:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917937-4096-sector-size","meta_rev_id":"cd1c901ee72dafee15681a5ae5de8ecf82075a06","_number":917937,"virtual_id_number":917937,"owner":{"_account_id":36976,"name":"Tudor Domnescu","display_name":"Tudor Domnescu","email":"tdomnesc@redhat.com","username":"tdomnesc"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-03 16:33:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2024-05-03 07:35:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-03 14:46:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-03 14:46:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-02 11:24:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-02 12:24:58.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2024-05-03 07:35:03.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"8b10ace0d62fcc1a613e4d6825d88aac1c1fdfa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36976,"name":"Tudor Domnescu","display_name":"Tudor Domnescu","email":"tdomnesc@redhat.com","username":"tdomnesc"},"date":"2024-05-02 09:34:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0df814ffbd721572a825965b4ae408f557ac4a90","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-05-02 09:43:46.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"28077a2e1de69529beb185870897c4b0b721884c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:24:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a67d5b9501c4092a1233e98e5a1d3d9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/089266616671436fb60021ac342e3e77 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/689ee6dd1bb348eaab947332bc44708e : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cb368551ecf4b2e805edf1392860619 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddfd4abd313b4db4a4297c60c58c3c61 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3039d4bcf6f14533ab5594478593317b : SUCCESS in 4m 53s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/775c2ccb1d094b80a29aefb7a19718a5 : SUCCESS in 1h 40m 11s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/a1bfa3e17735426ba3e7cf85db4a9867 : SUCCESS in 1h 39m 48s\n- ironic-lib-tox-codespell https://zuul.opendev.org/t/openstack/build/1e0a2147723a484088becb57f018d363 : SUCCESS in 4m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"35a2e48f106834427209bb01d2e0f289b6c93c9a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-05-02 12:24:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f916bb61cb19dc641aeb97090642cd854903d5e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36976,"name":"Tudor Domnescu","display_name":"Tudor Domnescu","email":"tdomnesc@redhat.com","username":"tdomnesc"},"date":"2024-05-02 13:33:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"049b292133edd3a270d13fac9a3ffe4f6c7d4848","author":{"_account_id":36976,"name":"Tudor Domnescu","display_name":"Tudor Domnescu","email":"tdomnesc@redhat.com","username":"tdomnesc"},"date":"2024-05-02 13:37:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8a49dce53896713c697d313a09f02413b64d8120","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 15:14:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/195e3eebe62f4e8fac1737adb74b42b7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c5672f004734adb98eb5c0b0863c0ef : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3042d59a926a4b4287b647a04b0cbfae : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b1606f4b2cf46198f619984c43120ce : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc7ddc0bd0ff49e28571bb71dfae287b : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a22fac83a324501bd45d0f7c51a28e1 : SUCCESS in 5m 20s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/93981b3eab1c412584c4b90ccccc218c : SUCCESS in 1h 35m 06s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/565f5c36eb754218803f844aeb3e43ad : SUCCESS in 1h 27m 01s\n- ironic-lib-tox-codespell https://zuul.opendev.org/t/openstack/build/560da9f6c6c745b7a4a8d2cfe7af1062 : SUCCESS in 2m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"57bb8f2f6b96cc99b482497fbedae251094bade2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-05-03 07:35:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a0835d8e62d66cf78608f046c4ee308131fc8ab2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-05-03 14:46:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"731a1d10176357bcf5c86f98ca6353dee751b37c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 14:47:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86694755a7e14fe325ea04fc38688b6918e34b24","author":{"_account_id":36976,"name":"Tudor Domnescu","display_name":"Tudor Domnescu","email":"tdomnesc@redhat.com","username":"tdomnesc"},"date":"2024-05-03 14:50:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e60c3a5f089df0145f0d12feb13d6f4a19adb310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 16:33:27.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/947772183a3a4b138f71b3fe1ad2c3cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adc84e3af6d94de2a44eb476ff883980 : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/694dc3b13ff2464e8f4b03120b5feecd : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84713df2b82743d3ac072af7ef8ddb04 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c705b5d5bd8b482c958150c8ff210358 : SUCCESS in 4m 42s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/d6014d408ac24475bcb20d29719960c6 : SUCCESS in 1h 41m 08s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/97c79b52a85a4cb5b77f4eea6b245535 : SUCCESS in 1h 04m 18s","accounts_in_message":[],"_revision_number":2},{"id":"f7ce70d50c7ebaddf9490fa005349dd40226ad2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 16:33:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"cd1c901ee72dafee15681a5ae5de8ecf82075a06","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 16:34:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14df306db8df4f3485519265f1a7aa53\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67080575a570429388fc48f562c18315 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"571b074a101725b0bdd4eeb29c9366d635808ad2","revisions":{"1d12b18265afb23da25e396d2c567418f390a598":{"kind":"REWORK","_number":1,"created":"2024-05-02 09:34:16.000000000","uploader":{"_account_id":36976,"name":"Tudor Domnescu","display_name":"Tudor Domnescu","email":"tdomnesc@redhat.com","username":"tdomnesc"},"ref":"refs/changes/37/917937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/37/917937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/37/917937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/37/917937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/37/917937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/37/917937/1"}}},"commit":{"parents":[{"commit":"11dc763c476a9761f1cebfacb78f1d4641b2dc90","subject":"Remove old excludes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/11dc763c476a9761f1cebfacb78f1d4641b2dc90"}]}],"author":{"name":"Tudor Domnescu","email":"tdomnesc@redhat.com","date":"2024-05-02 09:33:19.000000000","tz":120},"committer":{"name":"Tudor Domnescu","email":"tdomnesc@redhat.com","date":"2024-05-02 09:33:19.000000000","tz":120},"subject":"destroy_disk_metadata: support 4096 sector size","message":"destroy_disk_metadata: support 4096 sector size\n\nA sector size of 512 was assumed and hardcoded, causing dd to fail when\nit tried to write in chunks smaller than the sector size for disks with\n4096 bytes sectors. The size of GPT in sectors also depends on sector size.\n\nChange-Id: I9e6d1d7fb7342daa64c73244674c0bc3c420963f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/1d12b18265afb23da25e396d2c567418f390a598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/1d12b18265afb23da25e396d2c567418f390a598"}]},"branch":"refs/heads/master"},"571b074a101725b0bdd4eeb29c9366d635808ad2":{"kind":"REWORK","_number":2,"created":"2024-05-02 13:33:40.000000000","uploader":{"_account_id":36976,"name":"Tudor Domnescu","display_name":"Tudor Domnescu","email":"tdomnesc@redhat.com","username":"tdomnesc"},"ref":"refs/changes/37/917937/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/37/917937/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/37/917937/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/37/917937/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/37/917937/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/37/917937/2"}}},"commit":{"parents":[{"commit":"11dc763c476a9761f1cebfacb78f1d4641b2dc90","subject":"Remove old excludes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/11dc763c476a9761f1cebfacb78f1d4641b2dc90"}]}],"author":{"name":"Tudor Domnescu","email":"tdomnesc@redhat.com","date":"2024-05-02 09:33:19.000000000","tz":120},"committer":{"name":"Tudor Domnescu","email":"tdomnesc@redhat.com","date":"2024-05-02 13:32:31.000000000","tz":120},"subject":"destroy_disk_metadata: support 4096 sector size","message":"destroy_disk_metadata: support 4096 sector size\n\nA sector size of 512 was assumed and hardcoded, causing dd to fail when\nit tried to write in chunks smaller than the sector size for disks with\n4096 bytes sectors. The size of GPT in sectors also depends on sector size.\n\nChange-Id: I9e6d1d7fb7342daa64c73244674c0bc3c420963f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/571b074a101725b0bdd4eeb29c9366d635808ad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/571b074a101725b0bdd4eeb29c9366d635808ad2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
