)]}'
{"id":"openstack%2Fironic-lib~918257","triplet_id":"openstack%2Fironic-lib~unmaintained%2Fzed~Id26b247f5023c45ec22776274e569e1a2ab07951","project":"openstack/ironic-lib","branch":"unmaintained/zed","topic":"create-zed","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-07-17 19:35:49.000000000","reason":"\u003cGERRIT_ACCOUNT_10342\u003e replied on the change","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2024-07-17 22:25:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id26b247f5023c45ec22776274e569e1a2ab07951","subject":"Update .gitreview for unmaintained/zed, CI fixes","status":"MERGED","created":"2024-05-06 06:22:01.000000000","updated":"2024-07-17 22:27:51.000000000","submitted":"2024-07-17 22:25:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918257-create-zed","meta_rev_id":"2806bad8222265ea544aaa78c2c63367e4447d7a","_number":918257,"virtual_id_number":918257,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-17 22:25:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2024-07-17 15:55:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2024-07-17 15:56:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-06 07:03:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-06 09:20:46.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-06-12 09:01:19.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-07-16 17:17:53.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"a6ee198ecc72b64e54e99843f04ada221127dc61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-05-06 06:22:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9498f1438bfcb43d0f019c4a3f2512f42b6f203","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 07:03:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/afc61ce804cc4374822efd16a752bb5c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ea0f1f538344658975f72e22ba04952 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18feaca2508a43539b791b95cf465aae : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b356f662f85348a6ab8c9b118423ee20 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b222a1cbc08f4f51b334f0397c40d684 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/923f82d0bfd9491dbf45b5a3977781f1 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60588a92eb3e4f98999997a24dac52c3 : SUCCESS in 4m 44s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/86389a909d7943c782b50e0d74629174 : FAILURE in 25m 13s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/4aca36f066f64d68acac14b8f774b562 : FAILURE in 20m 36s","accounts_in_message":[],"_revision_number":1},{"id":"ac0fe3fc089bae563c91b186660d9d1c9d679e97","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-05-06 09:20:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"994334db8ebb70cc42330a79650602f5b60c8f3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 11:19:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6143f539fed4b909bb30cf5fc94319a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33cbd5e49d2a47a0aa51e737f049b1cb : SUCCESS in 3m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a7d178504114cb887ddb0bc8b45b40e : SUCCESS in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbfe9634b9d546899aa45c064b900ea9 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b38ef369056042b188a26e6e46b41b8b : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58e4c02344a047d1a3ee7a07688ea8f0 : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c7f62ca360e4a0d895fdc52e73d5e2c : SUCCESS in 6m 10s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/072d8389a6244b049a6b3c8a588aafdc : SUCCESS in 1h 38m 50s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/c0560ed8fa644b06b677b1045f045f73 : SUCCESS in 1h 20m 47s","accounts_in_message":[],"_revision_number":1},{"id":"cf920e53a367d761f3838cbf17c30018dfe40246","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 11:20:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9b7d37bbaaf86cc5f5ba15c0739ca1865ead8f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-06 11:49:24.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80ac1ffc945548f3882089fa91461ffd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6434209df25a4bb4b7e335ac6a231938 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/048716409d0746c995a51344792f9e9b : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec3e7812761e434ba063ef0936334800 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fc1deda9d4d43679cdc5628d4b65941 : SUCCESS in 6m 48s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/201b8cb803a049619b853309cea50e32 : FAILURE in 27m 43s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/38057b0540de450a85758beb65780d97 : FAILURE in 21m 24s","accounts_in_message":[],"_revision_number":1},{"id":"cfb0b0601f897a15eaa9a7aecd9ddec5a7628a1c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-06-12 09:01:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01f67ab6d3e5168aa3e0697c67861075b196385c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 10:43:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b1f0aa9f30e47469dcd1967f6a126d2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92284b8e50654eb898b379dae605a72b : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d4cd81ed3bd47a595bc5b6b4217a41f : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7215cfeae02413aa6ad612b09881ab0 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1453462e91141a783555dbd1453042f : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62000d2f914c4368bfd4c5baa4af377a : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1fb4a5df96947f89d764683c4513632 : SUCCESS in 5m 01s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/38e7d4eb3c214c47b5c233dc4ff85e63 : SUCCESS in 1h 37m 02s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/f2d7e6f09d22437794867c8c6d4b21fd : FAILURE in 25m 22s","accounts_in_message":[],"_revision_number":1},{"id":"aa9faf8b3accde667bcde13fd9d60896f3ddc0ef","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-16 17:17:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e6470da073d821df39ee9ba578793a02e09dd59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 19:18:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/803a89573cda4f0faba89734a84d684f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd32cb7ba03344bbb26f5709433bf1b4 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c66442a0f34913883170db7afdd791 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68c6ca1ed68e4e55986af2609108129d : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caa0fa3c83124b76b65aff45921e5210 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28aec46cef35488eae18da5506bff30f : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7229875c1bd14deab9eaabc2be0ea56c : SUCCESS in 5m 40s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/e61be7b965874abfbbcbb259e274f00d : SUCCESS in 1h 44m 51s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/0e13d55c590842828448cefab8be9ba5 : FAILURE in 30m 35s","accounts_in_message":[],"_revision_number":1},{"id":"37009fc189bfb9e26d2725c3072261d9059fa5ed","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-16 21:39:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d69cce7dda951dbb08c117742ee72dd00206503","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-16 21:40:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6dc423a3fda9812377f2483a29bc0056c08ca3e6","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-16 21:41:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"653ca4dce3b0f32c4738795424048a8c0be47ed2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-16 21:44:39.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f0b21f1760d51cc7316177b091e37886d86c102d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 23:33:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77515b7ffd7a4dbcaf90ce69afdd446a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a461508453048a2a0f170599c909e54 : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0938fce933c4789a07df56a472cf4c0 : SUCCESS in 2m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88a94113239c4c83a9f624d5f888486d : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3cd670283da4e6ea487ab4e6ba8bea0 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c0bc971922654c2c9e1922e2fd441437 : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d908666f0c4a249aced16765317468 : SUCCESS in 4m 04s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/876a2eaeef754cfb8d2ae04640eef935 : SUCCESS in 1h 35m 27s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/a62a9fca8c174088b770abf90eb646a2 : SUCCESS in 1h 41m 25s","accounts_in_message":[],"_revision_number":2},{"id":"0a3eab6db87e2b9f6cda1c9fcc2b8bc1647210ce","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-17 15:55:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ebcdcee559b07c79be26c013b87ae14628b6031d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-17 15:56:31.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9c92ef2e6d2f6e7b1092d65950138cc297d3381b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 15:57:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2b1ece567a5a6f3cbf72cce7d55fd81910c44b3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 17:07:35.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74018310553747f3853de063e6745c35\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb652a6db772443b94217cbcb9c760bd : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30be7e7d687a4b8b876c3b2fb3def0f5 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c23247bf60fd40849115c4c582f7e535 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efcdbccc61224ae998b36f4c94667258 : SUCCESS in 4m 58s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/d82e5e75a82748aeaa011cc245ef6851 : FAILURE in 40m 49s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/06e62f9f0a894fdeb29fe96b22d3f870 : SUCCESS in 1h 04m 27s","accounts_in_message":[],"_revision_number":2},{"id":"c86e7e54ed0cc6f868a15e029eb40c0300f6456b","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-17 19:35:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c5536cec5dfd2f2d5aa7eced38a08fe79d03c0c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 21:04:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df720b30a83842a4bf06c7085825160a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f19f8fb38b44e128f2f7445ec0e0a23 : SUCCESS in 2m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a90b2ef7024681bbf8cd1da38973e6 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af4dcd7423034a3996dffa615c5a6ef6 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0022fc74e1f4c3ebe48f6b603c42666 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a949e2a3a56340ce8e4ab8d1d3bd7563 : SUCCESS in 2m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/093b4224f19e42e7a2308329ce1ca4eb : SUCCESS in 4m 19s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/22069fe7e68e4131ab68e5e7fe21e494 : SUCCESS in 1h 14m 57s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/5b1024dc5e49458f8b6bdd4fa2b059ae : SUCCESS in 1h 22m 20s","accounts_in_message":[],"_revision_number":2},{"id":"79ce4a9ff0c9cb3f8a42a4ad7c0c889070f75ec3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 21:04:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"996fae5864a00aa934409ae2ad69c257044af0dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 22:25:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62634f3dda854c78a6b7bb43add8a73c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ada7a970acb48798fba5eb0527ee97e : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8b2b4b9b1f94f1dbcc4e27dc949cc5f : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c389af8eedf54472975c35a4b393be12 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b6c426bee4482f884f87c7b1b34273 : SUCCESS in 5m 32s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/9274304303ff40a29c093cf905a58a15 : SUCCESS in 1h 18m 09s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/2f7223c97c8740bdb07617195ad71655 : SUCCESS in 1h 07m 33s","accounts_in_message":[],"_revision_number":2},{"id":"bd220065a2f24c19a1983b5e188adc6f1e434626","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 22:25:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2806bad8222265ea544aaa78c2c63367e4447d7a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 22:27:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/925e086cca074902ae07ec8985ecb40b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66a9e7ac5afc498b9d6f2aa2b3e8e79c : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7d4fec679bb54c1768a7c400edbd1fac0eb1aa78","revisions":{"cca60dd432ccc75c13512bfafc4f0fc3a06ca878":{"kind":"REWORK","_number":1,"created":"2024-05-06 06:22:01.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/57/918257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/57/918257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/57/918257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/57/918257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/57/918257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/57/918257/1"}}},"commit":{"parents":[{"commit":"4ee0571c51d470bf3bc9d1adb4d76f8e8c0b170f","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/4ee0571c51d470bf3bc9d1adb4d76f8e8c0b170f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-05-06 06:22:00.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-05-06 06:22:00.000000000","tz":0},"subject":"Update .gitreview for unmaintained/zed","message":"Update .gitreview for unmaintained/zed\n\nChange-Id: Id26b247f5023c45ec22776274e569e1a2ab07951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cca60dd432ccc75c13512bfafc4f0fc3a06ca878"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/cca60dd432ccc75c13512bfafc4f0fc3a06ca878"}]},"branch":"refs/heads/unmaintained/zed"},"7d4fec679bb54c1768a7c400edbd1fac0eb1aa78":{"kind":"REWORK","_number":2,"created":"2024-07-16 21:44:39.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/57/918257/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/57/918257/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/57/918257/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/57/918257/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/57/918257/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/57/918257/2"}}},"commit":{"parents":[{"commit":"4ee0571c51d470bf3bc9d1adb4d76f8e8c0b170f","subject":"Update TOX_CONSTRAINTS_FILE for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/4ee0571c51d470bf3bc9d1adb4d76f8e8c0b170f"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-05-06 06:22:00.000000000","tz":0},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-07-16 21:43:55.000000000","tz":-420},"subject":"Update .gitreview for unmaintained/zed, CI fixes","message":"Update .gitreview for unmaintained/zed, CI fixes\n\nRequiring use of tinyipa images for testing here, since DIB image builds\nare failing for linux-reasons, and that\u0027s not what these ironic-lib jobs\nare testing.\n\nChange-Id: Id26b247f5023c45ec22776274e569e1a2ab07951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/7d4fec679bb54c1768a7c400edbd1fac0eb1aa78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/7d4fec679bb54c1768a7c400edbd1fac0eb1aa78"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
