)]}'
{"id":"openstack%2Fironic-lib~924267","triplet_id":"openstack%2Fironic-lib~stable%2F2023.1~Ic959d25838ba01e0c8a6a2882f151ed0c85affce","project":"openstack/ironic-lib","branch":"stable/2023.1","topic":"jayf-ci-test","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-07-17 16:26:34.000000000","reason":"\u003cGERRIT_ACCOUNT_10342\u003e replied on the change","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}},"hashtags":[],"change_id":"Ic959d25838ba01e0c8a6a2882f151ed0c85affce","subject":"noop change; testing ci","status":"ABANDONED","created":"2024-07-16 17:16:22.000000000","updated":"2024-10-10 13:39:01.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3dc0d54ddc453d6ac41be24ec20eee5615778fa6","_number":924267,"virtual_id_number":924267,"owner":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-17 17:58:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-16 18:57:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db17f79b897b21dec14fef9c56607446050007f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-16 17:16:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18f66c3ac46a7abd962a9adc275fb928df7f1949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 18:57:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16503097a36149fa9a2428d75366026c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/638c4bf94c2142028bf7a19ed62346ee : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cbc02eba56e4e97819edba757a88d81 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d960a66c18ff4e4fbe129f60557a2e09 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dabc765f5427448c8981010913352c8b : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d415de412964ff88a3679ac11e08e75 : SUCCESS in 7m 23s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/e5e5fe067ce94ffd8d7a0d84527cf6a2 : FAILURE in 25m 07s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/529d5631578649d99bae855430ad7515 : SUCCESS in 1h 28m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c1013440d57d08fdabded89aa298e15b6b9a0730","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-16 22:26:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"50d03f8bb0e46c465550114018d0e0d5824e3b00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 23:15:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea292bb1e3b54a4396f10a057b7bd3db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b87be1695fd243fe90df731c04c98009 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f20450c63a29416d80f9d1e3892cc141 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44511445910c4a0a81c41535eed1b2e4 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13665ee519924bc78df547dc375dcbd8 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27d2c72b69041c5b59e0b5f08adfc0e : SUCCESS in 6m 01s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/b9d72553db604017bd6805505c9892f3 : FAILURE in 40m 13s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/cbcc45ea39ea468f9bac905393c09c91 : FAILURE in 40m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9843f89db262b69cb891d3cb05fe16fc99a9bdc0","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-17 16:26:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bf6d4b182b993259ccb2c558f61c665e95a28e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 17:58:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7740c19d08f2425b8873fc6ab47b7926\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f52d29bc6a4f4731a0ace23294924af8 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bed82d9dc65a4ae687610a7f8e99cf50 : SUCCESS in 2m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4616b17e632d445f87f2fafea371796d : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1951dc6fb0b642db8ee34214f773cd9c : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f04457b0072749d194d185a874490ff2 : SUCCESS in 4m 50s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/aaec806d1dba4c1ea1e9de2eea455f81 : SUCCESS in 1h 30m 52s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/7ad53aa71c5f43609fafafa8d6e28987 : SUCCESS in 1h 16m 24s","accounts_in_message":[],"_revision_number":1},{"id":"3dc0d54ddc453d6ac41be24ec20eee5615778fa6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-10-10 13:39:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d6a5e56894854d98aa636ce2d6ba0243d5f57a5","revisions":{"6d6a5e56894854d98aa636ce2d6ba0243d5f57a5":{"kind":"REWORK","_number":1,"created":"2024-07-16 17:16:22.000000000","uploader":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"ref":"refs/changes/67/924267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/67/924267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/67/924267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/67/924267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/67/924267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/67/924267/1"}}},"commit":{"parents":[{"commit":"3a9cc010279184b39971d8b94c46803c267b0875","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/3a9cc010279184b39971d8b94c46803c267b0875"}]}],"author":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-07-16 17:16:19.000000000","tz":-420},"committer":{"name":"Jay Faulkner","email":"jay@jvf.cc","date":"2024-07-16 17:16:19.000000000","tz":-420},"subject":"noop change; testing ci","message":"noop change; testing ci\n\nChange-Id: Ic959d25838ba01e0c8a6a2882f151ed0c85affce\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/6d6a5e56894854d98aa636ce2d6ba0243d5f57a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/6d6a5e56894854d98aa636ce2d6ba0243d5f57a5"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
