)]}'
{"id":"openstack%2Fironic-lib~926719","triplet_id":"openstack%2Fironic-lib~stable%2F2023.1~I0c776e3fe5efa82b3220f573953bcac1678883f2","project":"openstack/ironic-lib","branch":"stable/2023.1","topic":"fix-utf8-surrogateescape-stable/2024.1-stable/2023.2-stable/2023.1","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2024-09-18 10:07:21.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}},"hashtags":["ironic-week-prio"],"change_id":"I0c776e3fe5efa82b3220f573953bcac1678883f2","subject":"Fix invalid UTF-8 characters in execute output","status":"MERGED","created":"2024-08-21 09:36:58.000000000","updated":"2024-09-18 12:50:30.000000000","submitted":"2024-09-18 12:49:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926719-fix-utf8-surrogateescape-stable/2024.1-stable/2023.2-stable/2023.1","cherry_pick_of_change":926718,"cherry_pick_of_patch_set":1,"meta_rev_id":"6887dc2c95eca7e8521c572a677bfc5c0b869edb","_number":926719,"virtual_id_number":926719,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-18 12:49:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"recommended":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"all":[{"value":1,"date":"2024-09-02 07:04:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":2,"date":"2024-09-03 23:17:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":1,"date":"2024-09-03 23:17:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-21 10:08:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-02 07:04:21.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"},{"updated":"2024-09-03 23:17:19.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"931c998b46ccac141debafaa0a9e39d30bed71e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-08-21 09:36:58.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"acc766b207ee186374c1dcb2b653400cb2ba9180","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 10:08:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/804cd823ccb44c7b81db31d3d5da3687\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9df051fa4ba4e57a6ced0d93e36b027 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0df19ba5b0744c8a69ae5952108036e : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69ff099087db487fbb9a0c63cc2b8d01 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4dbf4550f29240f7bd18637aa08b5fce : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8323944b7e3472cafe150e8c882d58e : SUCCESS in 6m 15s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/c74b5b6d85e94c71902d81b91099f88a : FAILURE in 26m 41s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/c1fc32761fb54999a4c48582ae044d0d : FAILURE in 24m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7a6d5250ba3b0eeaed46c2bd74d0f7567f39a878","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-08-21 14:19:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d71eb329bd05e8d6a21ecaec6a8d424ff441933b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 16:03:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a56ddd02b524619ac9f14e8427c9dd0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f47215ad16ec4ec6ad4429854b88b58c : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40f5c13a79ef41beaae6b8a263b286d8 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c47c8f17686d428fafaa785336a4cd86 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1da1275c2e944108b4f6763b9e393682 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35bb4dba27c043d483fdc77c7ee0f8d0 : SUCCESS in 5m 11s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/0e3e5e9541f843d7842f8b14079f77c0 : SUCCESS in 1h 36m 51s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/a2eda3c190c840a1a6f3763ba38a88c5 : SUCCESS in 1h 16m 29s","accounts_in_message":[],"_revision_number":1},{"id":"37315d8d1ef37b35515fdd972f85c9a659ba3be6","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-08-27 08:37:07.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"3887b3d3f25b9d6a225041cd929ffff5c5c62b8e","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2024-09-02 07:04:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3e9be9e77e72a06874482c4d9e0e474aa565ee1b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-09-03 23:17:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d39d1b5ef6e408664cc65ef1939f0edadb809bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 23:17:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0865e130895feaa63f63483b9da8d0b046d443eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 00:54:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c06106d3f05744bc9617c7e356196ec2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1afb1687703481597bb6aba6fc1d30c : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60409e63bb164a02b28cffc849715765 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/890641adabfe4ea38a6cb48d39131dcd : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2602423f66a847029de87584b19795b3 : SUCCESS in 4m 15s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/5c1dcd41c10e425e9eb5290ca6cd3e54 : FAILURE in 30m 07s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/90b95497ed2a4f2b9c2b7d61f66107ec : FAILURE in 25m 11s","accounts_in_message":[],"_revision_number":1},{"id":"9057d71b77a949cbc92d624b01a90a0e15198fb5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-09-18 10:07:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"600152f726020ef41bc1d2a129f7c4e77df27709","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 11:09:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/818f224ad5a44967bbab9b9341903388\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4172734cb12e45ca87de46af7f5331bd : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1658454a57464b5f9e6d7cbb1524a3c6 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72694c985e5d4fd9907243580c818068 : SUCCESS in 4m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60e7fccb5e274f15a657e4bed32d8290 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ed3dc72daa4c469634d8a356908fa4 : SUCCESS in 3m 59s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/52b3df0770de473782aa7314dede6e84 : SUCCESS in 1h 00m 44s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/7b51f10a47724f25b9e7e9232f44c494 : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ef34eda4e8091876dfd09bdd9b84bb705f4c8210","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 11:09:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"935f03a6790df5d12bde532cc36d0713e0e0836b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 12:49:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d412e309f614f738c78f1bbe084219d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2c902d37f2b4cccb2f5556225202626 : SUCCESS in 2m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b70e5fc47294961955bd8a3a4e1a8bf : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bf605553583439b828029d347647941 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74cda2024ff6425aaa95a8d813a5205d : SUCCESS in 5m 13s\n- ironic-lib-uefi-ipmi-src https://zuul.opendev.org/t/openstack/build/dd17c171225b4667adb80d9731c27323 : SUCCESS in 1h 34m 35s\n- ironic-lib-bios-ipmi-src https://zuul.opendev.org/t/openstack/build/b44be561630248d1a495f65164416231 : SUCCESS in 1h 27m 27s","accounts_in_message":[],"_revision_number":1},{"id":"5f8296ba4e0482295c5f4e5ad32cca82f72881ff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 12:49:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6887dc2c95eca7e8521c572a677bfc5c0b869edb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 12:50:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71a65a3aeb7040ecbd199dadcf8ac749\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bb3e06c53ea4cd5b04ac2146c0c0cc2 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9cf477f9079b91f571b276e1c666a466828b0ec","revisions":{"e9cf477f9079b91f571b276e1c666a466828b0ec":{"kind":"REWORK","_number":1,"created":"2024-08-21 09:36:58.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/19/926719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/ironic-lib","ref":"refs/changes/19/926719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/19/926719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/19/926719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/ironic-lib refs/changes/19/926719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/ironic-lib refs/changes/19/926719/1"}}},"commit":{"parents":[{"commit":"3a9cc010279184b39971d8b94c46803c267b0875","subject":"Update TOX_CONSTRAINTS_FILE for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/3a9cc010279184b39971d8b94c46803c267b0875"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-08-09 10:01:35.000000000","tz":120},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-08-21 09:36:58.000000000","tz":0},"subject":"Fix invalid UTF-8 characters in execute output","message":"Fix invalid UTF-8 characters in execute output\n\nIt\u0027s rare but the output of commands called by execute can\ncontain invalid UTF-8 characters.\nWhile the execute command does not care about the output\nitself and will finish execution correctly, transposing\nthat into string (for example when logging the output)\nwill raise a Unicode exception.\nWhen that happens we should escape the invalid characters\nand log a trimmed output.\n\nCloses-Bug: 2071935\nChange-Id: I0c776e3fe5efa82b3220f573953bcac1678883f2\n(cherry picked from commit 1ca3c8cf1a90ccaed619df8b4ba97dbf9e386b98)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/e9cf477f9079b91f571b276e1c666a466828b0ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/ironic-lib/commit/e9cf477f9079b91f571b276e1c666a466828b0ec"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
